)]}'
{"id":"openstack%2Fcliff~966914","triplet_id":"openstack%2Fcliff~master~I9ee1e73ca7858457cd0e3343b0d191c967ad9d09","project":"openstack/cliff","branch":"master","topic":"conflict-resolution","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-11-24 12:06:30.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I9ee1e73ca7858457cd0e3343b0d191c967ad9d09","subject":"Implement conflict resolution","status":"MERGED","created":"2025-11-12 18:41:07.000000000","updated":"2025-11-25 13:45:37.000000000","submitted":"2025-11-25 13:44:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"966914-conflict-resolution","meta_rev_id":"cac03567cccf3dd9c9f7965e4ca72ea58ed07898","_number":966914,"virtual_id_number":966914,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-25 13:44:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-25 11:46:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-25 11:46:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-12 19:02:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4f9778eecb1c6f128df0fd9ca6048af5297c439e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-12 18:41:07.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"328b8164b32cb9170a089338b63d518cfa8a3b3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-12 19:02:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/abd310b930704e83877f3ee1ecda82c4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/df0c78fe46d1413fa3da378cbd08da1b : FAILURE in 18m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/044347e5833a4bb8b9ee0efe103bae47 : FAILURE in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d75350fae27f45f3891115b847edd130 : FAILURE in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f4f49e74d9f646d7bd52978db73842cd : FAILURE in 4m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/908d1a5b04d147079d231121bfa528c9 : FAILURE in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6a6c0ecc44454f488ad65f4356637923 : FAILURE in 3m 05s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/16e4ff36319f4e35899085047e843623 : FAILURE in 4m 47s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/05fcbf3957734c0cbecbcf6c8bdac9ef : FAILURE in 4m 33s","accounts_in_message":[],"_revision_number":1},{"id":"f936137bdf44df584a5f7cd441f6ac82c98f9027","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-17 17:43:44.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"84fde6ac18878e9d0fd9dbae8bd66d5572ad8d7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-17 17:56:39.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5f407af05cac4419882fec5c1112683a\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ac1016d763fc412fafaedd8830f9bb50 : FAILURE in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/753bd60e41f34356a80e46fed54a1bb8 : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f5ba2c071b349369a69b7e89cf40b6e : FAILURE in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/326c1579621a4c04bdb7fee97bf0c359 : FAILURE in 3m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f5a85270d6944b15a7bac32c180e240f : FAILURE in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/816e3cf228e0426a994612c51fbc7e5c : FAILURE in 4m 10s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/fbc6a2f502a94869aa0259bb8821bffe : FAILURE in 2m 55s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/d2c8666a758040ef9f8b8bc597a9493f : FAILURE in 4m 40s","accounts_in_message":[],"_revision_number":2},{"id":"cec47ccc6063ec578a236134cbf194e52a81a2a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-21 11:34:42.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"c83f47055b8e9dd9297e52cab537e0a9fbcb8031","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-21 11:46:56.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/fad214af48f8425c984753fafb8efe9d\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/f270dc811a054d5ea1e9e9f6a0f3787e : FAILURE in 8m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7c71ecefac73400d984da70bf2a70bd2 : SUCCESS in 3m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/87ae52ad20d64addb47c8fb9c8a25097 : FAILURE in 2m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ca17b3e371e94cebba2bd14e3a6ca5d1 : FAILURE in 2m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5aa61c1891024470a2b3650d96f9c904 : FAILURE in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a927002bc204b3788f68627bd58bb24 : SUCCESS in 4m 12s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/33c4eba7044d42989ce3d4e46963b07c : FAILURE in 4m 12s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/49be2496bb4b46f097bec78f2a7285ca : FAILURE in 3m 47s","accounts_in_message":[],"_revision_number":3},{"id":"2d9bc9f3756169175bc239fbcb87209c0aef3a00","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-24 12:06:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"12d88cc57e07ce76c5cfcd45ca7ead2d6a642542","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 13:17:22.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/642413d8291c49aaaea3241e56bd6df4\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/0745bf117a2744cca15e43cedaced62f : SUCCESS in 1h 10m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/299906951ca743579b191b8bb70d1822 : SUCCESS in 2m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/71a2f8cfabf9495a82439079aebed9f9 : SUCCESS in 2m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44b21265e8ab4bd59dc9fcf678bec1e6 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4cae1341939d440d9171f6fcd8d9c39a : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aff5e5ee4bce4ff6aca47daba731dad5 : SUCCESS in 1m 47s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/66c846ce6f6040298141c6e57164d7b0 : SUCCESS in 3m 03s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/a9ea17da8b8342f49b032d9e8ab69da1 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":3},{"id":"8dd13debd11590272ad9ad180ece33d9e34c62fc","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-25 11:46:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"7b5230e4fabc0d7950c247e8c5a064b3c9f9ec23","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 11:47:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"95f3aaf9aa8d509f6a65c1c4e87afe25a405417a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 13:44:07.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91e668badbad4f2d88f3f123a8864cda\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/6675025240484993af4072e47b4d7c2c : SUCCESS in 1h 47m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efd22f3b337b4ea4bc47e469bbacbf89 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7f1c42e6f17e4d4cbdfdfd2fee9a2824 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b0df74706e5e4e3ab931dcddb3ce1590 : SUCCESS in 3m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb9cfb92d965453092e7d83ecff0864c : SUCCESS in 2m 44s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/75c5773443844fbbb36ccf26f91bff2b : SUCCESS in 3m 20s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/b81f1a5ba15a4054b82036761e48e743 : SUCCESS in 5m 13s","accounts_in_message":[],"_revision_number":3},{"id":"b7bd21cb5e2a58d921c8222c1f36c07aace78a68","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 13:44:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"cac03567cccf3dd9c9f7965e4ca72ea58ed07898","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-25 13:45:37.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9191905ffa54c2c8db688e677abf1ac\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f91264246a294e7684bcef7adf5eb6ab : SUCCESS in 42s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"a2da89a8a2d65e16ffb4f3f991d47da3763cf3a4","revisions":{"e2dcd3b3b83fea128913eac4fc44694633716b85":{"kind":"REWORK","_number":1,"created":"2025-11-12 18:41:07.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/966914/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/14/966914/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/14/966914/1"}}},"commit":{"parents":[{"commit":"8ef4b118f5276735019c28b498cd5b6f276f1833","subject":"WIP: typing: Fixups for typed stevedore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/8ef4b118f5276735019c28b498cd5b6f276f1833"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-07 15:30:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-12 18:40:47.000000000","tz":0},"subject":"Implement conflict resolution","message":"Implement conflict resolution\n\nChange-Id: I9ee1e73ca7858457cd0e3343b0d191c967ad9d09\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/e2dcd3b3b83fea128913eac4fc44694633716b85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/e2dcd3b3b83fea128913eac4fc44694633716b85"}]},"branch":"refs/heads/master"},"fe5760fdd1a421a9007bcfb9c55e1f99e2416d04":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-11-17 17:43:44.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/966914/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/14/966914/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/14/966914/2"}}},"commit":{"parents":[{"commit":"b5c023f030325e49b2b4d1e5f062b5a0a87b80ad","subject":"WIP: typing: Fixups for typed stevedore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b5c023f030325e49b2b4d1e5f062b5a0a87b80ad"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-07 15:30:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-17 17:43:34.000000000","tz":0},"subject":"Implement conflict resolution","message":"Implement conflict resolution\n\nChange-Id: I9ee1e73ca7858457cd0e3343b0d191c967ad9d09\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fe5760fdd1a421a9007bcfb9c55e1f99e2416d04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/fe5760fdd1a421a9007bcfb9c55e1f99e2416d04"}]},"branch":"refs/heads/master"},"a2da89a8a2d65e16ffb4f3f991d47da3763cf3a4":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-11-21 11:34:42.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/14/966914/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/14/966914/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/14/966914/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/14/966914/3"}}},"commit":{"parents":[{"commit":"b33c023f9b71e5c70bfa410d6eedffe721f1915b","subject":"typing: Fixups for typed stevedore","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/b33c023f9b71e5c70bfa410d6eedffe721f1915b"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-07 15:30:49.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-21 11:34:18.000000000","tz":0},"subject":"Implement conflict resolution","message":"Implement conflict resolution\n\nChange-Id: I9ee1e73ca7858457cd0e3343b0d191c967ad9d09\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a2da89a8a2d65e16ffb4f3f991d47da3763cf3a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a2da89a8a2d65e16ffb4f3f991d47da3763cf3a4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
