)]}'
{"id":"openstack%2Fcliff~968815","triplet_id":"openstack%2Fcliff~master~I436944774d7cefc196f7bbdabc6319c9783fe2f2","project":"openstack/cliff","branch":"master","topic":"conflict-resolution","hashtags":[],"change_id":"I436944774d7cefc196f7bbdabc6319c9783fe2f2","subject":"Deprecate CommandManager namespace argument","status":"MERGED","created":"2025-11-28 15:14:51.000000000","updated":"2025-11-28 18:36:00.000000000","submitted":"2025-11-28 18:35:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"968815-conflict-resolution","meta_rev_id":"4c9e9d496f74768bd8f7df43f4b2b145bd2d0880","_number":968815,"virtual_id_number":968815,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-11-28 18:35:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-11-28 16:10:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-11-28 16:10:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-11-28 17:00:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"48e42666cd4ac9f26aea98316e67bfdb62e187f8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-28 15:14:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34cbf265f6f8d345ca25c123466b8c6d68c83cf6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-28 15:44:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9b6d10a8754e38100c01023c8c236101196efb13","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-11-28 16:10:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"53487581787033912a01f47e156ef96f56942432","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 17:00:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/45aa539c5c054e08872166a136bd4f30\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/5d8030267b244f13896b9ae4a3280aec : SUCCESS in 1h 06m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd2cae3a93654d648e51095b744c4d30 : SUCCESS in 4m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4d94dbfec33d4a29bdbf62241ba66329 : SUCCESS in 2m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/938bf949cb0f4c6dbb86c656aa480f13 : SUCCESS in 3m 06s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/614501484eb44bf4b3383279b3d7e83a : SUCCESS in 6m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b385facf02ae4ae39ec640ce41a8ca70 : SUCCESS in 3m 06s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/01d9ef08106348339ca8896eb15e2e0d : SUCCESS in 3m 00s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/25082bdc98824bb0924b03dcafb2ff98 : SUCCESS in 4m 47s","accounts_in_message":[],"_revision_number":2},{"id":"4a5bd3532a0c6388295fa997375a0a869b00b634","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 17:17:51.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7c6ff10a5f646deed878ceee36b4868a067537e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:35:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df9ca94a26f649ab9ccfa3c9a6068714\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/d3f78677e9ab41e7b127d5cbb3c996e2 : SUCCESS in 1h 16m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/032b4fb156354f1585ecce8b20d03726 : SUCCESS in 3m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf900c4f665749b79d8b046e645f10b1 : SUCCESS in 2m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/101427f5f2be41b2a524121fb4045ea8 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea09c39c2c474029998711a15e80fdc1 : SUCCESS in 4m 28s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/94c01d059fb740748fcebba2c114ed2a : SUCCESS in 1m 56s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/321a59088107428583f42f3e4f002a33 : SUCCESS in 3m 56s","accounts_in_message":[],"_revision_number":2},{"id":"2c886bb7fc0ee841511c257ac2e37e4fe8980025","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:35:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"4c9e9d496f74768bd8f7df43f4b2b145bd2d0880","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-28 18:36:00.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6dd58a7124b413d921eef8a1d3c8ccf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/217402ceb3fc4edea5c0daa7b0baed28 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0d82750388b2a82d5b67d6d3a81ee150d234d4d2","revisions":{"0ba0d36556b8047628677a1bd345e8a0da3cbdcc":{"kind":"REWORK","_number":1,"created":"2025-11-28 15:14:51.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/968815/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/15/968815/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/15/968815/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/15/968815/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/15/968815/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/15/968815/1"}}},"commit":{"parents":[{"commit":"ebb9c8c48eb76c9e74cd360f8cffc97d2fd566ed","subject":"Warn on duplicate commands in the same namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/ebb9c8c48eb76c9e74cd360f8cffc97d2fd566ed"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-28 15:05:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-28 15:14:25.000000000","tz":0},"subject":"Deprecated CommandManager namespace argument","message":"Deprecated CommandManager namespace argument\n\nChange-Id: I436944774d7cefc196f7bbdabc6319c9783fe2f2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0ba0d36556b8047628677a1bd345e8a0da3cbdcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0ba0d36556b8047628677a1bd345e8a0da3cbdcc"}]},"branch":"refs/heads/master"},"0d82750388b2a82d5b67d6d3a81ee150d234d4d2":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2025-11-28 15:44:47.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/15/968815/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/15/968815/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/15/968815/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/15/968815/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/15/968815/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/15/968815/2"}}},"commit":{"parents":[{"commit":"a0eee2b08db2bc2db315d1cc70df2eebcd9690c0","subject":"Warn on duplicate commands in the same namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a0eee2b08db2bc2db315d1cc70df2eebcd9690c0"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-28 15:05:07.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-11-28 15:44:35.000000000","tz":0},"subject":"Deprecate CommandManager namespace argument","message":"Deprecate CommandManager namespace argument\n\nChange-Id: I436944774d7cefc196f7bbdabc6319c9783fe2f2\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0d82750388b2a82d5b67d6d3a81ee150d234d4d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0d82750388b2a82d5b67d6d3a81ee150d234d4d2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
