)]}'
{"id":"openstack%2Fcliff~970670","triplet_id":"openstack%2Fcliff~master~Id52ba2684f3799cc4df4229c9326294a33fa3ad9","project":"openstack/cliff","branch":"master","topic":"typing","hashtags":[],"change_id":"Id52ba2684f3799cc4df4229c9326294a33fa3ad9","subject":"Run mypy from tox","status":"MERGED","created":"2025-12-11 18:33:12.000000000","updated":"2025-12-11 21:48:46.000000000","submitted":"2025-12-11 21:47:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"970670-typing","meta_rev_id":"7c6eee15b45d1ca7e7a9ff05b68ef9cffdb07e41","_number":970670,"virtual_id_number":970670,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-12-11 21:47:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2025-12-11 19:16:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2025-12-11 19:16:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-12-11 20:31:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0b8de62f71f9eb950e5b4b0ceb3fa069f2cef494","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-11 18:33:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b07850d29db845ff1a8be39f59f34af26904b06d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2025-12-11 19:16:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb597af61a4626e9849bf5d650b491f75b61d8de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 20:31:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7f35ab43bff4cc69e9b004e3463186a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/caf5dd9d6b1f4d66acfe83cde35a6ad4 : SUCCESS in 2m 28s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e980e2b4a65c40ca840e0ed6c71fec95 : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8b1b11b62060423cbe873bf802f6776c : SUCCESS in 1h 49m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea4beecb611542f4a44d74c94239ae8c : SUCCESS in 3m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc1be73733fd47c381ea564d5acdb470 : SUCCESS in 1m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3229aa218a244ba18d41924fffc5a2f1 : SUCCESS in 4m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/16fdcb1bfcc346c09de5d33f959ed62b : SUCCESS in 6m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6468ee1e674049f0a5068ee20d25c7e8 : SUCCESS in 3m 07s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/512b629f2b5d4fc79bfcb079f2c2769f : SUCCESS in 4m 48s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/8bd3bb16b79740f6a8df9e07b58e39f6 : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"e78024d55ea61e68c41bf8e6cbfbe96d1f242642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 20:31:31.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2cd40e85f1cbc76dedefff6e0312c6b7f8324acf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:47:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/850d2090f89b4b04975c059acad2bd6d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/732eb5d103f14d11957ee5138b9d97f5 : SUCCESS in 5m 25s\n- requirements-check https://zuul.opendev.org/t/openstack/build/eb047dff020a4866803df6196000deda : SUCCESS in 2m 14s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a3ad01113eca4ef891256be7d80c0128 : SUCCESS in 1h 16m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76be7330bf4b4c4a83a4806640c08ce5 : SUCCESS in 3m 39s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/affcca6525dd4d6c969e0de13c91210a : SUCCESS in 2m 55s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/aede936f26da45d3beef27fa4c554e6e : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f707fe1a8154e25a463300a26fd6af6 : SUCCESS in 3m 01s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/01c108a2fddd4d52803f38bad539e0b9 : SUCCESS in 2m 50s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/65fc68566cfb4414a387e6039352b877 : SUCCESS in 4m 55s","accounts_in_message":[],"_revision_number":1},{"id":"1e6a2036285062458565a2dccec33be261fe6ec6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:47:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7c6eee15b45d1ca7e7a9ff05b68ef9cffdb07e41","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 21:48:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fb0a5acdd31342ca80eb81542a91eceb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb97edbcb2cc47b098a185054fdadf81 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"90b29a8a980843597f7badb507d4da0e0b34b3b3","revisions":{"90b29a8a980843597f7badb507d4da0e0b34b3b3":{"kind":"REWORK","_number":1,"created":"2025-12-11 18:33:12.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/70/970670/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/70/970670/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/970670/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/970670/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/70/970670/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/70/970670/1"}}},"commit":{"parents":[{"commit":"0d82750388b2a82d5b67d6d3a81ee150d234d4d2","subject":"Deprecate CommandManager namespace argument","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0d82750388b2a82d5b67d6d3a81ee150d234d4d2"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-12-11 18:26:00.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2025-12-11 18:32:04.000000000","tz":0},"subject":"Run mypy from tox","message":"Run mypy from tox\n\nThis avoids the need to duplicate our dependency list in multiple\nplaces and allows us to take advantage of tox\u0027s dependency management\ninfrastructure, to ensure we always get the latest and greatest version\nof a package allowed by upper-constraints.\n\nChange-Id: Id52ba2684f3799cc4df4229c9326294a33fa3ad9\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/90b29a8a980843597f7badb507d4da0e0b34b3b3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/90b29a8a980843597f7badb507d4da0e0b34b3b3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
