)]}'
{"id":"openstack%2Fcliff~972186","triplet_id":"openstack%2Fcliff~master~Ibd938439348af5f7e932fb63796d1b5fa53eaa35","project":"openstack/cliff","branch":"master","topic":"typing","hashtags":[],"change_id":"Ibd938439348af5f7e932fb63796d1b5fa53eaa35","subject":"Add typing classifier","status":"MERGED","created":"2026-01-05 15:05:18.000000000","updated":"2026-01-05 18:31:12.000000000","submitted":"2026-01-05 18:30:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"972186-typing","meta_rev_id":"e5d99aeb9843f78b16e97da36d0992b76c98b851","_number":972186,"virtual_id_number":972186,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-01-05 18:30:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-01-05 15:23:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-01-05 15:23:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-01-05 16:59:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c6edc500c0ba49604c5d9a8b4e24cf51a51e6ea9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 15:05:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"20c81fafb348cedbdf19d958ab33dbd8d04f5699","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-01-05 15:23:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"eb1dc926bccda319012093b1f13637eef9102827","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1faa04c4d87940579b564b17f023cb54\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/03e022b5a6ec4b9eacd896147ba03b13 : SUCCESS in 2m 10s\n- requirements-check https://zuul.opendev.org/t/openstack/build/722bb17be8b64537b2f6d19bd8a3fe23 : SUCCESS in 1m 59s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/877944b6d04c46aba9286295e9d83d48 : SUCCESS in 1h 43m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8fe7bd31ea0454988db832b603bcf66 : SUCCESS in 2m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d7718819846b428794a33187f13491ee : SUCCESS in 2m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/aed33631ae8b40c1ab8964f661d6c3c3 : SUCCESS in 3m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/673c671a5c124eaf8f1f52d98c5fc42a : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b00d25a76845018307f1044dea0310 : SUCCESS in 2m 46s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/82644eff6ec14523872bdeec9c757366 : SUCCESS in 3m 30s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/f5ec46b06efc45e78dc6781aebfa470f : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":1},{"id":"8f1ecf89e757ef66455974622bcfb9cf276191e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 16:59:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"750dcd8e168f600f090d663cf804f6e902c0b705","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:30:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6746f93efee34bc093bf9f6f92bc834d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c4d0c285606149bd9436de6b5a0a90d7 : SUCCESS in 2m 31s\n- requirements-check https://zuul.opendev.org/t/openstack/build/389b7b4ed8534ceb8803d9c579ca8c2f : SUCCESS in 2m 01s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/8ab333879a004cb5a2b99c0dda832b20 : SUCCESS in 1h 22m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/745e0d712f9d40169814106a580201ff : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c41d24042be34eeea4775faea380db6d : SUCCESS in 2m 43s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b55bd582a25f40d3b3d816eaa95528e9 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3cdf78fcd87404b8302ca2ec8042470 : SUCCESS in 3m 25s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/00fc9f40a412458e816a55619e877da1 : SUCCESS in 2m 37s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/58bf6b2f6d644b6aa116cc88e82f4d33 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"1f057bf4582e85a0b1d6424802949f342c0b7ea0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:30:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e5d99aeb9843f78b16e97da36d0992b76c98b851","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-01-05 18:31:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8fb3d00b5ffd4baaa55a3163233c13d4\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1442d6c166d2465987c04bbc9a4c11e4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4752f90da9f9a03178a1100902ec212d130891e2","revisions":{"4752f90da9f9a03178a1100902ec212d130891e2":{"kind":"REWORK","_number":1,"created":"2026-01-05 15:05:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/86/972186/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/86/972186/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/86/972186/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/86/972186/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/86/972186/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/86/972186/1"}}},"commit":{"parents":[{"commit":"2bf314464b10d5cfc862d6a0bb067070736d2198","subject":"Remove use of ABCMeta for formatters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/2bf314464b10d5cfc862d6a0bb067070736d2198"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-05 15:05:12.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-01-05 15:05:12.000000000","tz":0},"subject":"Add typing classifier","message":"Add typing classifier\n\nThis is parsed by PyPI.\n\nChange-Id: Ibd938439348af5f7e932fb63796d1b5fa53eaa35\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4752f90da9f9a03178a1100902ec212d130891e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4752f90da9f9a03178a1100902ec212d130891e2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
