)]}'
{"id":"openstack%2Fcliff~978755","triplet_id":"openstack%2Fcliff~master~I75933ec1ffc2dc8d57d4105c5683892ddbf50bff","project":"openstack/cliff","branch":"master","topic":"cmd2-3x","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-03-05 10:52:11.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I75933ec1ffc2dc8d57d4105c5683892ddbf50bff","subject":"Fix compatibility with cmd2 3.1.0+","status":"MERGED","created":"2026-03-03 12:47:31.000000000","updated":"2026-03-05 14:48:14.000000000","submitted":"2026-03-05 14:47:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978755-cmd2-3x","meta_rev_id":"6df28c176cebda7da32d0169d9613532ead6c3ab","_number":978755,"virtual_id_number":978755,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-05 14:47:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-05 12:00:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-05 12:00:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-03 14:13:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1fe8192d9d9181ae6f31c7c10b477a704fd28b8d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-03 12:47:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"733c7fff2433142d78a6e6f87b803bd881b93af2","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-03 13:14:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"16a8feadfc687327155c4cddf8a199721d55a8a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-03 14:13:52.000000000","message":"Patch Set 1: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/14865d0af700478b99522566e3400057\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/02f5d37f89ff4d9fbef1ae25e40eec38 : SUCCESS in 2m 30s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4932078cbdcf465e94f742db7459f19b : SUCCESS in 1h 24m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90229f2398fd4e2d8e53080b1249b370 : FAILURE in 3m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0611e3ea2f644dc8ad0bcbd3c08121b : SUCCESS in 2m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26cffb2703fd49848a57281949ba6f70 : SUCCESS in 3m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4e07bbce2b30448186985f3cf76f8c50 : SUCCESS in 3m 16s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/8b983850c34446f383075a293f970e59 : FAILURE in 6m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3971b736e8544d8d8271fa466356efd3 : SUCCESS in 4m 01s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/3885a0e662bd4d56bf4725d30cb5a1bb : SUCCESS in 3m 40s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/f059c4a0202346ed9e881bb7a2a1da7c : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":1},{"id":"5bbc01186e3e14daf683f0131e796457f498c5ed","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 10:52:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe3a28c470d8abbbbf3afc9ae1945227496de7fe","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 12:00:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"29586c265933f29ccb8fd2979fcb6a8b3ab17902","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 12:51:45.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/69821ae7757e40bf9ce06092ddd13989\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/e2553839a2a74748ab3e955e7bc43115 : SUCCESS in 2m 04s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/002468180f914ff08cb55757b085eeb7 : SUCCESS in 1h 53m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/721c461ae1444f98998dcc141850d687 : SUCCESS in 3m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/863939754a8e4e92a978d5e4792b9df1 : SUCCESS in 4m 27s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db5b0e9ea1374701be4a0ab5bd311b7e : SUCCESS in 3m 01s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/eb3c4df6494245169ecbf5ae07521fa3 : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/43e85154caf44cb1a69afb410899a695 : FAILURE in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c99a682a7b4646e48fd53a530658740e : SUCCESS in 3m 04s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/185cf95af99649e38adab6dc3c476097 : SUCCESS in 3m 18s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/ca83865c758a401493b5627f0fb1af06 : SUCCESS in 4m 18s","accounts_in_message":[],"_revision_number":1},{"id":"b86c517788b07f1fb6dda5da9e15d1ed565e0c4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 12:52:08.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"75927faa6945a5556ecc0d1dd22ff42e73bc85bf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 14:47:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4248273b04b343ba8ca5962bcb2be48e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/ead072decf384785a55db48a039011f6 : SUCCESS in 1m 33s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/7a0db36b7ecf40839300ff2706ba241a : SUCCESS in 1h 50m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/84427c80b5564b9298004e791db0b31e : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d273a97a0ef24e97ba05b1f80deacf2c : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/6530461629cd4aa19769ded909568de5 : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e054601d80345709a97a1ad0af032fc : SUCCESS in 2m 44s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/21995436e5914b1a9629da01b9d18ad7 : SUCCESS in 2m 52s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/6560e6c3ad8147eb968cd4cb7d799572 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"e579ab904d014c1cab02427a6fb7c8d58a0a4254","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 14:47:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6df28c176cebda7da32d0169d9613532ead6c3ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 14:48:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/717a0f1f2ef24119a65cee7b8e14ac62\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2d8b42a57a449dda85ec63ca4f1571b : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"29fa87c0687063fdb3fd9c58e58c44aca7bce8a8","revisions":{"29fa87c0687063fdb3fd9c58e58c44aca7bce8a8":{"kind":"REWORK","_number":1,"created":"2026-03-03 12:47:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/55/978755/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/55/978755/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/55/978755/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/55/978755/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/55/978755/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/55/978755/1"}}},"commit":{"parents":[{"commit":"391261c849c994ca2d3f42926497e633047ed8c7","subject":"Fix cliff sphinxextention for Python 3.14","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/391261c849c994ca2d3f42926497e633047ed8c7"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-03 12:10:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-03 12:46:52.000000000","tz":0},"subject":"Fix compatibility with cmd2 3.1.0+","message":"Fix compatibility with cmd2 3.1.0+\n\ncmd2 3.1.0 has removed the dependency on the stdlib cmd library [1]. For\nour use cases, the largest change is that we no longer have the\n\u0027completenames\u0027 helper, which was defined in `cmd.Cmd` rather than\n`cmd2.Cmd`. However, upon further inspection is seems this helper was\nnever actually used. It was not called directly in cmd2 2.x and was only\nused in cmd in two locations: `Cmd.complete` [2] and `Cmd.complete_help`\n[3]. cmd2 2.x entirely overrode the former [4] and deleted the latter,\nmeaning we were not actually using any of that code. Therefore we can\njust delete the helper and any tests for same and suffer no ill effects.\nTo be safe though, we bump our cmd2 minimum to 3.0.0.\n\nWhile here, we also need to fix some type hints to account for the cmd2\nbump along with a related autopage bump.\n\n[1] https://github.com/python-cmd2/cmd2/releases/tag/3.1.0\n[2] https://github.com/python/cpython/blob/v3.10.20/Lib/cmd.py#L274\n[3] https://github.com/python/cpython/blob/v3.10.20/Lib/cmd.py#L287\n[4] https://github.com/python-cmd2/cmd2/blob/2.7.0/cmd2/cmd2.py#L2192-L2294\n[5] https://github.com/python-cmd2/cmd2/blob/2.7.0/cmd2/cmd2.py#L3770-L3772\n\nChange-Id: I75933ec1ffc2dc8d57d4105c5683892ddbf50bff\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/29fa87c0687063fdb3fd9c58e58c44aca7bce8a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/29fa87c0687063fdb3fd9c58e58c44aca7bce8a8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
