)]}'
{"id":"openstack%2Fcliff~978994","triplet_id":"openstack%2Fcliff~stable%2F2026.1~I75933ec1ffc2dc8d57d4105c5683892ddbf50bff","project":"openstack/cliff","branch":"stable/2026.1","topic":"cmd2-3x","hashtags":[],"change_id":"I75933ec1ffc2dc8d57d4105c5683892ddbf50bff","subject":"Fix compatibility with cmd2 3.1.0+","status":"MERGED","created":"2026-03-05 12:02:30.000000000","updated":"2026-03-05 12:21:50.000000000","submitted":"2026-03-05 12:20:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978994-cmd2-3x","meta_rev_id":"d19b96ec32c3048272dd27d6ddecc5417a59f917","_number":978994,"virtual_id_number":978994,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-05 12:20:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-05 12:02:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-05 12:02:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-05 12:12:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"c518c0f2f0b48f726b0616ae206c181bb1f5888d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 12:02:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ca2f31216de7d9daa7d3a278befef3a2a98fa767","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 12:02:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a569df1e2c76ff78ce9414cdea8103d86a626cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 12:12:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5009c39fe67d423fb55d86dc0c8614e1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/82ab5574fb8745fa8dd909689f6d1cd2 : SUCCESS in 3m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78a7ee46ae8f4d8da44db731f608a456 : SUCCESS in 4m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/97d4ff2a8cc7495eb06edd6270b8b768 : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/13052882c9eb4fb69c92912e293993f3 : SUCCESS in 2m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad1ad23d156c4bcb87204f7d9f94385e : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ee10dcd45fb5992baedb4b56ce0a549b422d3406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 12:12:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"9fde6353a310d6d937ffbef24b6c101ed2c9a559","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 12:20:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cb43825c16f24160af0dc0b81ec10445\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bac5ca7c1c6e46919c238542774aa1f5 : SUCCESS in 2m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/64b900ea309e4bd18a45fdae948752d1 : SUCCESS in 2m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f47301d43aca45449822a22f4604ab33 : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c2acd438fbd43328f3f03d3c36cba7d : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":1},{"id":"7c460ce9245fb7ac8cc40025637aa0c7d13c88f5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 12:20:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"d19b96ec32c3048272dd27d6ddecc5417a59f917","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 12:21:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dcbf7ccb2b3b4b5c97447b0d26d60a1b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23c3ace8907d49bcaef90db8069f5f41 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd88434b240473f5f9400294ae9c854367d1e0ae","revisions":{"cd88434b240473f5f9400294ae9c854367d1e0ae":{"kind":"REWORK","_number":1,"created":"2026-03-05 12:02:30.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/94/978994/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/94/978994/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/94/978994/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/94/978994/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/94/978994/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/94/978994/1"}}},"commit":{"parents":[{"commit":"0c1dbcbae6259fe80a6c9d4d44a38ae4445c1e48","subject":"Revert \"Deprecate CommandManager namespace argument\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/0c1dbcbae6259fe80a6c9d4d44a38ae4445c1e48"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-03 12:10:54.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-05 12:02:19.000000000","tz":0},"subject":"Fix compatibility with cmd2 3.1.0+","message":"Fix compatibility with cmd2 3.1.0+\n\ncmd2 3.1.0 has removed the dependency on the stdlib cmd library [1]. For\nour use cases, the largest change is that we no longer have the\n\u0027completenames\u0027 helper, which was defined in `cmd.Cmd` rather than\n`cmd2.Cmd`. However, upon further inspection is seems this helper was\nnever actually used. It was not called directly in cmd2 2.x and was only\nused in cmd in two locations: `Cmd.complete` [2] and `Cmd.complete_help`\n[3]. cmd2 2.x entirely overrode the former [4] and deleted the latter,\nmeaning we were not actually using any of that code. Therefore we can\njust delete the helper and any tests for same and suffer no ill effects.\nTo be safe though, we bump our cmd2 minimum to 3.0.0.\n\nWhile here, we also need to fix some type hints to account for the cmd2\nbump along with a related autopage bump.\n\n[1] https://github.com/python-cmd2/cmd2/releases/tag/3.1.0\n[2] https://github.com/python/cpython/blob/v3.10.20/Lib/cmd.py#L274\n[3] https://github.com/python/cpython/blob/v3.10.20/Lib/cmd.py#L287\n[4] https://github.com/python-cmd2/cmd2/blob/2.7.0/cmd2/cmd2.py#L2192-L2294\n[5] https://github.com/python-cmd2/cmd2/blob/2.7.0/cmd2/cmd2.py#L3770-L3772\n\nChange-Id: I75933ec1ffc2dc8d57d4105c5683892ddbf50bff\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n(cherry picked from commit 29fa87c0687063fdb3fd9c58e58c44aca7bce8a8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/cd88434b240473f5f9400294ae9c854367d1e0ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/cd88434b240473f5f9400294ae9c854367d1e0ae"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
