)]}'
{"id":"openstack%2Fcliff~979027","triplet_id":"openstack%2Fcliff~master~Ie9d6f62005523c697f409f978b09510ff681a634","project":"openstack/cliff","branch":"master","topic":"trivial","hashtags":[],"change_id":"Ie9d6f62005523c697f409f978b09510ff681a634","subject":"tests: Drop use of testscenarios","status":"MERGED","created":"2026-03-05 16:10:18.000000000","updated":"2026-03-05 19:03:39.000000000","submitted":"2026-03-05 19:02:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979027-trivial","meta_rev_id":"4131a967b73e32df0826b8e07f35d4fea0544b64","_number":979027,"virtual_id_number":979027,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-05 19:02:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-05 16:26:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-05 16:26:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-05 17:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"acf3f96ef36079b4d462b20ce13f4b4f96ea5eae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 16:10:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fdc3c464b20e4eb0b11b473f2d6757ad350113ef","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 16:10:30.000000000","message":"Topic set to trivial","accounts_in_message":[],"_revision_number":1},{"id":"5839856045130062da933dc2ce60baa0eda1d805","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 16:26:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"206c6ae5b170a9cbc095ee5c5be3543dce8b815f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 17:12:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/65003f470b644868be824f83e1fe766a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/2d9be3c9293c41138fdd62ba0ca60359 : SUCCESS in 1m 52s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/677babd7ac87449fad547a41ad5c08cc : SUCCESS in 1h 00m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/201cc0cec2554a97912967e37cba8a59 : SUCCESS in 2m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e6f5cb4a0d57458bbdd1ac8918a09623 : SUCCESS in 3m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad1cf4d472f048af9b74f224b0cfcbdb : SUCCESS in 2m 14s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5442113a91584d3ebef8cf33bf8724ea : SUCCESS in 1m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a71963ca601946c3b7b848a6c5cbafac : FAILURE in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c54ba4338a1b4fc69d4fcb19d1afa1e4 : SUCCESS in 2m 06s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/a20c2059288f44e1a9377833a35a7780 : SUCCESS in 2m 06s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/9eac243b5ee647348bdd595b11e59e76 : SUCCESS in 6m 44s","accounts_in_message":[],"_revision_number":1},{"id":"b20a0c6c0017ba7310ddd12a211b70c5956cc16b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 17:12:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a50682317c0bcff5a6d4f72af4104349b5a46649","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 19:02:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a9fe0ee05c074c79ad0cf694c0988089\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5fe89296fd5a414f925e267630c140a2 : SUCCESS in 2m 27s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/9071b485fc744f9bae828e3094a66736 : SUCCESS in 1h 44m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf28db300b242c79e8c3dda97249e00 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d03581f0d75409c8be44db42ddce13f : SUCCESS in 1m 54s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/61d50dff94334fee9317d8b770a1717f : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/793c4fd4140848eeaff4c5c7475f9b81 : SUCCESS in 2m 55s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/dff9906a12f4433697480d8711940743 : SUCCESS in 3m 34s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/e9818bdaa42d47c29daafe071e8ffa30 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"97b6aa16bf84aef58398ce600d28c39e38cf2d5b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 19:02:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4131a967b73e32df0826b8e07f35d4fea0544b64","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 19:03:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ea72c77797cc4f3190c64a377f340c8b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a29c23dbd24745be900114830078586a : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4ec63e5c2061e0abccda9edccecab1b7e6781f95","revisions":{"4ec63e5c2061e0abccda9edccecab1b7e6781f95":{"kind":"REWORK","_number":1,"created":"2026-03-05 16:10:18.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/27/979027/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/27/979027/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/27/979027/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/27/979027/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/27/979027/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/27/979027/1"}}},"commit":{"parents":[{"commit":"29fa87c0687063fdb3fd9c58e58c44aca7bce8a8","subject":"Fix compatibility with cmd2 3.1.0+","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/29fa87c0687063fdb3fd9c58e58c44aca7bce8a8"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-05 16:07:53.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-05 16:09:08.000000000","tz":0},"subject":"tests: Drop use of testscenarios","message":"tests: Drop use of testscenarios\n\nI have never really liked this library, and its use is no longer\nnecessary now that unittest.TestCase provides the subTest helper.\n\nWhile we\u0027re here, we also remove coverage from requirements (it\u0027s over\nrequired for the coverage job) and organise dependencies alphabetically.\n\nChange-Id: Ie9d6f62005523c697f409f978b09510ff681a634\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4ec63e5c2061e0abccda9edccecab1b7e6781f95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4ec63e5c2061e0abccda9edccecab1b7e6781f95"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
