)]}'
{"id":"openstack%2Fcliff~979030","triplet_id":"openstack%2Fcliff~master~I1f5ca2d391f67a064bd307db0ac0c1268a6bd81b","project":"openstack/cliff","branch":"master","topic":"trivial","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-04-27 10:32:59.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"I1f5ca2d391f67a064bd307db0ac0c1268a6bd81b","subject":"test: Use context manager-style assertRaises","status":"MERGED","created":"2026-03-05 16:25:56.000000000","updated":"2026-06-27 02:51:54.000000000","submitted":"2026-06-27 02:50:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"979030-trivial","meta_rev_id":"45141cfd6736c6fedc42544face1a1b606b19344","_number":979030,"virtual_id_number":979030,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-27 02:50:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":0,"date":"2026-06-27 00:58:09.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":2,"date":"2026-04-26 16:04:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-06-27 00:58:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-05 17:54:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-04-26 16:04:35.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"65d9c20bc84ce5c9ed2d9e8f0b31cad4d378bf0c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 16:25:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3befa627f941fa679c841940d443e4969b0a69d","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-05 16:26:49.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fd96f6e75ee51dd72eedbb039ec4de9eff7d45b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-05 17:54:29.000000000","message":"Patch Set 1: Verified-1\n\n(22 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/db7394560b1f4c3ea994ed76ae610e08\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/0b3f9f1ab43a424f9dad508eccf852b6 : SUCCESS in 1m 31s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/ef2b37a1171d4045a21bb7e4eca4d23e : SUCCESS in 1h 27m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b96ce50cb5f4275a050ab3166f600b4 : FAILURE in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0845503742d340578891f73e77d5fd59 : SUCCESS in 2m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7765c94200d34290a175537db6d9defa : SUCCESS in 2m 22s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/917d592c825141b68682074369bdf190 : SUCCESS in 6m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/cea4d6d31f3243fc8d2864574882464a : FAILURE in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/af073e39ae18494ea8b72d851e4c82a8 : SUCCESS in 2m 14s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/8ac92cd700af4b8693aab6c6a037c193 : SUCCESS in 2m 26s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/546bd2a9db13419f8b55ff2816763c66 : SUCCESS in 4m 56s","accounts_in_message":[],"_revision_number":1},{"id":"5a2deafb48b7d662e0dacc71979874d99b39abb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-09 08:27:57.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"bb83ef11ce0f69d3eb43b75736c7bdd1eec28bda","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-09 08:28:22.000000000","message":"Patch Set 2: -Code-Review","accounts_in_message":[],"_revision_number":2},{"id":"b0fcf7634c9689edfb993b9fee01324c0e3351ef","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-09 08:28:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ca219599a9e51a74dbc154230a1df2ce52ff6faf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-04-09 10:35:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/355075771c334a1c87ee7d6ff03f961d\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3556daf0f7d94a4d9004182d5dd82cbb : SUCCESS in 2m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/4dc54a7eef834127a4fb4e2211b730e0 : SUCCESS in 2h 00m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38175f7be78042cb90642b76b991b038 : SUCCESS in 4m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/35f40037a7fd4838bc821a4f7b663dd6 : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c37c9ac211684bca984dee814c13cbd9 : SUCCESS in 4m 04s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/ceaea999724f49e2be196e23ebf65266 : SUCCESS in 3m 19s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b70c6e9f304d425096809c95d2604f78 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b5f5140e744215b2c94507c8e3d62f : SUCCESS in 2m 43s\n- cliff-tox-py312-neutronclient-tip https://zuul.opendev.org/t/openstack/build/23df33336fca4ef5b64d584181b8f250 : SUCCESS in 4m 32s\n- osc-tox-py312-tips https://zuul.opendev.org/t/openstack/build/c954b8feeb8544b5b09f06cedcd8addf : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":2},{"id":"19ab03cc494b2660d801f1f315ef972f651a0fa5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2026-04-26 16:04:35.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f463473d0b65872e58564bb2103be79401c66b15","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-04-27 10:32:59.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"64703602b9a2408ae2674f8cbea86f73051b233a","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-06-27 00:58:09.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"e6a4e521c3142da2d36502172f9c2303b8e3a030","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 00:58:40.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"9a21f9717b38cf598da9e4f85a690df4804c89de","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 02:50:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0de06505fd24f578d22999078f417d4\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/1da1540cefe84e3f80895855bfbfa06d : SUCCESS in 2m 21s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/19e1c57ec0d745c9979066bdafd400cd : SUCCESS in 1h 46m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/30f273faa2204db8b11fa6836e645d09 : SUCCESS in 2m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec3365d001fd4e8b8ce5feb8c542c324 : SUCCESS in 1m 57s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/734f350983f945e0b912a5ea26fab609 : SUCCESS in 3m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/12ccd4f20be64c0bbaa2943bdb2fdadb : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99a2c208096d476294cb7eeee502e316 : SUCCESS in 2m 54s\n- cliff-tox-py313-neutronclient-tip https://zuul.opendev.org/t/openstack/build/f02941fbb13b455aabfe4f9c24b0e6ec : SUCCESS in 3m 03s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/83dddafa3a2043b79e23f8aad0e26d5f : SUCCESS in 6m 26s","accounts_in_message":[],"_revision_number":2},{"id":"a3ea7a623548bc22155f47710426f2c330f09940","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 02:50:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"45141cfd6736c6fedc42544face1a1b606b19344","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-27 02:51:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e08b00b840bb432989cd1b1283ea5f9c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293409ff06724543b0e0980d6bc1652c : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a386a3758d4824d8bc076bd986c22263ee537eb8","revisions":{"6439b902cf6d2adf3129908754a111a0574c39cd":{"kind":"REWORK","_number":1,"created":"2026-03-05 16:25:56.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/30/979030/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/30/979030/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/30/979030/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/30/979030/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/30/979030/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/30/979030/1"}}},"commit":{"parents":[{"commit":"4ec63e5c2061e0abccda9edccecab1b7e6781f95","subject":"tests: Drop use of testscenarios","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4ec63e5c2061e0abccda9edccecab1b7e6781f95"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-05 16:13:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-05 16:13:40.000000000","tz":0},"subject":"test: Use context manager-style assertRaises","message":"test: Use context manager-style assertRaises\n\nChange-Id: I1f5ca2d391f67a064bd307db0ac0c1268a6bd81b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6439b902cf6d2adf3129908754a111a0574c39cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/6439b902cf6d2adf3129908754a111a0574c39cd"}]},"branch":"refs/heads/master"},"a386a3758d4824d8bc076bd986c22263ee537eb8":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2026-04-09 08:27:57.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/30/979030/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/30/979030/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/30/979030/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/30/979030/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/30/979030/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/30/979030/2"}}},"commit":{"parents":[{"commit":"9cb939000413b78cd789550f2d899ad03771f4e5","subject":"typing: Remove additional use of private _argparse module","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9cb939000413b78cd789550f2d899ad03771f4e5"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-05 16:13:40.000000000","tz":0},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-04-09 08:27:57.000000000","tz":0},"subject":"test: Use context manager-style assertRaises","message":"test: Use context manager-style assertRaises\n\nChange-Id: I1f5ca2d391f67a064bd307db0ac0c1268a6bd81b\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a386a3758d4824d8bc076bd986c22263ee537eb8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/a386a3758d4824d8bc076bd986c22263ee537eb8"}]},"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
