)]}'
{"id":"openstack%2Fcliff~982757","triplet_id":"openstack%2Fcliff~stable%2F2026.1~I9bc601df02ede10e3fabceaef80d43f737872969","project":"openstack/cliff","branch":"stable/2026.1","hashtags":[],"change_id":"I9bc601df02ede10e3fabceaef80d43f737872969","subject":"typing: Remove additional use of private _argparse module","status":"MERGED","created":"2026-03-31 09:37:20.000000000","updated":"2026-03-31 13:27:26.000000000","submitted":"2026-03-31 13:26:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982757","meta_rev_id":"4a7be29f0d39f8c621061b33d0e4e079f18b9a62","_number":982757,"virtual_id_number":982757,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-31 13:26:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-03-31 09:38:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-03-31 09:38:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-31 11:26:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e2491902388227fcabd66034dc1c0bd08e40fdda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:37:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"70d75921eacb08056c01e97ed5f784eed78cd176","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-03-31 09:38:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"714aad2154bb7cd961d1e83bbaa6c0dae94b1aa7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:26:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3180b6c6f318485e8dfb38c9c8c091fc\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/2fda7196abf4446285aa6fd9b03345a5 : SUCCESS in 1h 44m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/760cef6256f947c2b32d4cfd3e3c3911 : SUCCESS in 3m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/90ec689121ea4a42a5ab6c1294e42fdf : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9abdb8436dfc4159901ce9809dcb8a19 : SUCCESS in 3m 32s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e527d3a15fc44d9b8173044e1a0f6b37 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/732b0572a7dc4a40952dcab3ca225985 : SUCCESS in 3m 28s","accounts_in_message":[],"_revision_number":1},{"id":"a3b30a42f3a9a2336473c2ab804261c03f086e12","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 11:27:26.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8da3bc6745b40a069d75e9de3e00e6155de70b0c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 13:26:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d97c6b70ce2145b3b7a2938f7da5b174\n\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b10e98806ce649ffbdaf43950964ee52 : SUCCESS in 1h 53m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76c5884d2c064f0c99743874e7b1c6cf : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1869b820fd749419ddf37c47abdab0b : SUCCESS in 2m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/17b754e7ebe943e69213bb45170a9b2e : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4280fd13faba405aa6efb663cbe82a6e : SUCCESS in 4m 58s","accounts_in_message":[],"_revision_number":1},{"id":"98c7216567185a9b1dbce390d5c6c169794a3992","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 13:26:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4a7be29f0d39f8c621061b33d0e4e079f18b9a62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-31 13:27:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31ab32852f244d2293d73a8bdb191428\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bf20f9a148674829bb5afe194dcf7ce0 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"44c0ee98e44dc6d73f2de295be2bf7423404cddb","revisions":{"44c0ee98e44dc6d73f2de295be2bf7423404cddb":{"kind":"REWORK","_number":1,"created":"2026-03-31 09:37:20.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/57/982757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/57/982757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/982757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/982757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/57/982757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/57/982757/1"}}},"commit":{"parents":[{"commit":"580bdc051b55182909b4bdc9d8f2a65e2e4ae9ce","subject":"typing: Broaden return type of Command.get_parser","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/580bdc051b55182909b4bdc9d8f2a65e2e4ae9ce"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-29 10:29:45.000000000","tz":120},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-03-31 09:33:08.000000000","tz":120},"subject":"typing: Remove additional use of private _argparse module","message":"typing: Remove additional use of private _argparse module\n\nLike I58b0be9d39ec6f6091ba2013436c11256b4befc7 but for cliff.app.App and\ncliff.hooks.CommandHook.\n\nConflicts:\n  cliff/app.py\n\nNOTE(stephenfin): Conflicts are due to the absence of change\nIc4a0384a700f5d63824940b2a215671839167450, which we don\u0027t want to\nbackport.\n\nChange-Id: I9bc601df02ede10e3fabceaef80d43f737872969\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n(cherry picked from commit 9cb939000413b78cd789550f2d899ad03771f4e5)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/44c0ee98e44dc6d73f2de295be2bf7423404cddb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/44c0ee98e44dc6d73f2de295be2bf7423404cddb"}]},"branch":"refs/heads/stable/2026.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
