)]}'
{"id":"openstack%2Fcliff~989011","triplet_id":"openstack%2Fcliff~master~Ic5ed3a51548a9f361205bf0ecaeb6e68bb4b2f58","project":"openstack/cliff","branch":"master","topic":"drop-py310","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-18 19:47:09.000000000","reason":"\u003cGERRIT_ACCOUNT_15334\u003e replied on the change","reason_account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}},"hashtags":[],"change_id":"Ic5ed3a51548a9f361205bf0ecaeb6e68bb4b2f58","subject":"Drop support for Python 3.10","status":"MERGED","created":"2026-05-18 17:01:06.000000000","updated":"2026-05-18 23:09:29.000000000","submitted":"2026-05-18 23:07:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"989011-drop-py310","meta_rev_id":"c68706a14e26fe643c8afda66ce57095bb07acba","_number":989011,"virtual_id_number":989011,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-05-18 23:07:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":2,"date":"2026-05-18 19:47:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":1,"date":"2026-05-18 19:47:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-05-18 17:53:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"bc7d752da6548f798f91f77d7f3b93656b441274","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-18 17:01:06.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ff637254d031f66f757d32335f718d77ed1e7f44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 17:53:36.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/70455ef7e6854d81966c9136a648039e\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8fbaa53052324c1e92ac403a65a0ff20 : SUCCESS in 2m 23s\n- requirements-check https://zuul.opendev.org/t/openstack/build/1485933ac0ba490e8cbcf1138afcdedb : SUCCESS in 2m 18s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/40d87d42149b4a18a68de82da8ebc76f : POST_FAILURE in 43m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/04f7fe2dd5bf40329a1bb437461f26f7 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c95dfbbf9a7a42bdba47616f39084db6 : SUCCESS in 4m 58s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa42bbf810e94b6f965f3eb66335b73f : SUCCESS in 4m 59s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/1302daacd0ed4d149141d62e85612da5 : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89b8a855165a498dacdf7538b1c1974d : SUCCESS in 4m 50s\n- cliff-tox-py313-neutronclient-tip https://zuul.opendev.org/t/openstack/build/98d34a43fa1c4eb8b3fe39ef77372cea : SUCCESS in 3m 18s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/f1f359e1cd6a46008404161350492928 : SUCCESS in 5m 09s","accounts_in_message":[],"_revision_number":1},{"id":"2feed82c846c23a1e10c7ad14758b95c16d5325f","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-18 19:47:09.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f8f01a3d1eecb6dd32695ce62ff43d5b5ef8e2a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 21:48:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63586259a325461cb820159aa5bdcef8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/11ee364150b24dbba8d22a6584410ec3 : SUCCESS in 3m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ee85b80591654872a4907189b012dc60 : SUCCESS in 3m 07s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/a5888f6d7f294eccb210fdcc4ad23a83 : SUCCESS in 1h 57m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b8b738dd1a9472c9b9d46ebe2bef28d : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9668b09f0ddc41759a88896dd49769a1 : SUCCESS in 2m 56s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/0b1c789d6d1844f6975b1c641ba347fe : SUCCESS in 3m 06s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/9eb6fce442e84443880bc3e4c770e9e3 : SUCCESS in 7m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/371e582275eb4af4a30f5e3372aeaa5f : SUCCESS in 4m 27s\n- cliff-tox-py313-neutronclient-tip https://zuul.opendev.org/t/openstack/build/ea23cf1d998e44d7a7cecbb9eac357e2 : SUCCESS in 3m 18s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/dc542e425a3c43a4bfbc1d3469c2d50d : SUCCESS in 4m 17s","accounts_in_message":[],"_revision_number":1},{"id":"df3b7403b9568f1120f8dbd75800d3c9f58c41a3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 21:49:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"17d11b4878d1bcd3b877b647263ee99234d7e7c3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:07:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89637b95134e43d09c8cdc3f207bc1bc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0f907d65a01949d6ba63e2f81559e5b9 : SUCCESS in 1m 29s\n- requirements-check https://zuul.opendev.org/t/openstack/build/23b385bd15d6455f8d2ad108343bc281 : SUCCESS in 2m 08s\n- tempest-full-py3 https://zuul.opendev.org/t/openstack/build/b8c7792c59554279919b195f2e5336fd : SUCCESS in 1h 16m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f599beee9f945d5bfb9a0d46b1d0a46 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10feacfdebfa46da84dc5b7590e2b69a : SUCCESS in 3m 08s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c569baec5f51406fac43dc80a32fa1a1 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7295fba1d98346acbde56050e86d8bd4 : SUCCESS in 3m 01s\n- cliff-tox-py313-neutronclient-tip https://zuul.opendev.org/t/openstack/build/8c88612bb94d4ecc8b973f768c9cc1ec : SUCCESS in 2m 29s\n- osc-tox-py313-tips https://zuul.opendev.org/t/openstack/build/520dbf8f637045bc8d7926e0a5194943 : SUCCESS in 4m 31s","accounts_in_message":[],"_revision_number":1},{"id":"319845dededc2d95feccf69cbd6baa7c3df4823c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:07:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c68706a14e26fe643c8afda66ce57095bb07acba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-18 23:09:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a1ff4a4c15c04e12b316f8c541b6c5cf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/13fecd4b444441009f695c62ac609994 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4b0ecb973e08063d0af484c229cb4076be34d601","revisions":{"4b0ecb973e08063d0af484c229cb4076be34d601":{"kind":"REWORK","_number":1,"created":"2026-05-18 17:01:06.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/11/989011/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cliff","ref":"refs/changes/11/989011/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/989011/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/989011/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cliff refs/changes/11/989011/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cliff refs/changes/11/989011/1"}}},"commit":{"parents":[{"commit":"9e12d0980849f17c4fa691138f478f406d7488f3","subject":"Fix command name alignment broken by ANSI color codes in help output","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/9e12d0980849f17c4fa691138f478f406d7488f3"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-05-18 16:59:52.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2026-05-18 17:00:55.000000000","tz":60},"subject":"Drop support for Python 3.10","message":"Drop support for Python 3.10\n\nChange-Id: Ic5ed3a51548a9f361205bf0ecaeb6e68bb4b2f58\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4b0ecb973e08063d0af484c229cb4076be34d601"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cliff/commit/4b0ecb973e08063d0af484c229cb4076be34d601"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
