)]}'
{"id":"openstack%2Fcloudkitty-tempest-plugin~739319","triplet_id":"openstack%2Fcloudkitty-tempest-plugin~master~I482ac98bc56f0e3cfb8b767f47649da11ed1afab","project":"openstack/cloudkitty-tempest-plugin","branch":"master","topic":"migrate-to-focal","hashtags":[],"change_id":"I482ac98bc56f0e3cfb8b767f47649da11ed1afab","subject":"migrate testing to ubuntu focal","status":"ABANDONED","created":"2020-07-04 17:57:28.000000000","updated":"2020-07-09 20:19:30.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"aaaf205ae5981b5eab7e87d07fbdd422ff470007","_number":739319,"virtual_id_number":739319,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2020-07-05 02:54:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-07-05 02:54:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"dd0c4897e7388cebb418d31fae2fcb0655c54d60","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-04 17:57:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74fbde7cec47c34b29e7756691aad79ea9d73fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-04 18:50:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cba010d99f8d472e8f716ad54cd10cbc : FAILURE in 3m 27s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/b3a366d06295435c88ae4631cc7347e1 : SUCCESS in 37m 46s\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/bbc8fa9a86e444a7ab1fa30d905095bc : SUCCESS in 37m 48s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/da5e91732dc245f58ed774331417300c : SUCCESS in 39m 22s","accounts_in_message":[],"_revision_number":1},{"id":"f791352b04efdbd9da8ff215420cc27acdbc7cf4","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-05 02:07:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"71bb3de0badd58cf5e923089eefd03e1d3c93500","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-05 02:13:20.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"7847ce99fb3d7045200660547a7b5a5d7d812460","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-05 02:54:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/752e1334a6d8425c858a4161c7ad27c6 : SUCCESS in 3m 43s\n- requirements-check https://zuul.opendev.org/t/openstack/build/ba685e4dcb9444ceb652fbf4eef5bca4 : SUCCESS in 2m 20s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/00c1dadeee1548c0bda4451980eb69da : SUCCESS in 36m 19s\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/f55feb03abf9473984f4e9b4d5ab8667 : SUCCESS in 30m 54s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/184070b328d04ec4bfbf896e60c01d69 : SUCCESS in 32m 15s","accounts_in_message":[],"_revision_number":3},{"id":"bd9f988e5d6b974b0d08aa3b2acb842ec7d79816","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-07-09 20:19:30.000000000","message":"Abandoned\n\ndoing in https://review.opendev.org/#/c/740361/","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e5aa1ba543fd4a78d07312a08b16d513cb8004c4","revisions":{"ee4d09bf9e54cd0542a2e01e21aa66715fab31b6":{"kind":"REWORK","_number":1,"created":"2020-07-04 17:57:28.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/739319/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty-tempest-plugin","ref":"refs/changes/19/739319/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/1"}}},"commit":{"parents":[{"commit":"44b8c4ed91f875e54ab3e4f9591196710f17ef3b","subject":"More cleanup for py2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/44b8c4ed91f875e54ab3e4f9591196710f17ef3b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:55:20.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:57:27.000000000","tz":-300},"subject":"DNM: testing gate on ubuntu focal","message":"DNM: testing gate on ubuntu focal\n\nThis tests:\n- integration tests jobs\n- unit tests jobs\n- functional tests jobs\n- docs jobs\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\nStory: #2007865\nTask: #40180\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/ee4d09bf9e54cd0542a2e01e21aa66715fab31b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/ee4d09bf9e54cd0542a2e01e21aa66715fab31b6"}]},"branch":"refs/heads/master"},"c2b6e3b7dd4ca7aff9d156b4c5b7299008da84ea":{"kind":"REWORK","_number":2,"created":"2020-07-05 02:07:52.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/739319/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty-tempest-plugin","ref":"refs/changes/19/739319/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/2"}}},"commit":{"parents":[{"commit":"44b8c4ed91f875e54ab3e4f9591196710f17ef3b","subject":"More cleanup for py2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/44b8c4ed91f875e54ab3e4f9591196710f17ef3b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:55:20.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-05 02:06:11.000000000","tz":-300},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nStory: #2007865\nTask: #40178\n\nCloses-Bug: #1886296\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/c2b6e3b7dd4ca7aff9d156b4c5b7299008da84ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/c2b6e3b7dd4ca7aff9d156b4c5b7299008da84ea"}]},"branch":"refs/heads/master"},"e5aa1ba543fd4a78d07312a08b16d513cb8004c4":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-05 02:13:20.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/19/739319/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty-tempest-plugin","ref":"refs/changes/19/739319/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty-tempest-plugin refs/changes/19/739319/3"}}},"commit":{"parents":[{"commit":"44b8c4ed91f875e54ab3e4f9591196710f17ef3b","subject":"More cleanup for py2 drop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/44b8c4ed91f875e54ab3e4f9591196710f17ef3b"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-04 17:55:20.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2020-07-05 02:13:17.000000000","tz":0},"subject":"migrate testing to ubuntu focal","message":"migrate testing to ubuntu focal\n\nAs per victoria cycle testing runtime and community goal[1]\nwe need to migrate upstream CI/CD to Ubuntu Focal(20.04).\n\nFixing:\n- bug#1886296\nBump the pyflakes to 2.1.1 as min version to run pep8 jobs\non py3.8 which is default python vesion in ubuntu focal.\n\nDepends-On: https://review.opendev.org/#/c/734700/\n\nStory: #2007865\nTask: #40180\n\nCloses-Bug: #1886296\n[1] https://governance.openstack.org/tc/goals/selected/victoria/migrate-ci-cd-jobs-to-ubuntu-focal.html\n\nChange-Id: I482ac98bc56f0e3cfb8b767f47649da11ed1afab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/e5aa1ba543fd4a78d07312a08b16d513cb8004c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty-tempest-plugin/commit/e5aa1ba543fd4a78d07312a08b16d513cb8004c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
