)]}'
{"id":"openstack%2Fcloudkitty~1000891","triplet_id":"openstack%2Fcloudkitty~stable%2F2025.1~I11e8b9025ef2492be80e5c46e367153d02712330","project":"openstack/cloudkitty","branch":"stable/2025.1","attention_set":{"35511":{"account":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"last_update":"2026-08-13 20:16:51.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I11e8b9025ef2492be80e5c46e367153d02712330","subject":"Add storage backend migration command","status":"NEW","created":"2026-08-13 19:25:15.000000000","updated":"2026-08-13 20:16:51.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a08560d75f691c77f0fdeb58c0ac2ae5d8639761","_number":1000891,"virtual_id_number":1000891,"owner":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-08-13 20:16:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-08-13 20:16:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e6207f27720f7baaf719bc078bfc5df82a9d238d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"date":"2026-08-13 19:25:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a08560d75f691c77f0fdeb58c0ac2ae5d8639761","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-08-13 20:16:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d991d09060344349d38ff9c8f898c87\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3cf2fcd4519842f0ab42501e34b08610 : SUCCESS in 3m 53s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0465f93bed2c42998a50b861ccd5bcfd : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dea4ffd2085a48de8ac0a5b027ae06f7 : SUCCESS in 1m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3072b5cd39f4adea887660d9d688517 : SUCCESS in 2m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7eb305f216e4dfd96d8351b250e1b96 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ff050bcd6be485aa0228582ea69854e : SUCCESS in 2m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/895f687d26d84ca6889f890cdf220b2e : SUCCESS in 2m 55s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/fc4b1d7c79ee4279891280f2294a1c16 : SUCCESS in 14m 30s\n- cloudkitty-tempest-full-v2-storage-influxdb-jammy https://zuul.opendev.org/t/openstack/build/4854fee5e71b46208dd2641965cb3587 : FAILURE in 12m 57s\n- cloudkitty-tempest-full-v2-storage-influxdb-v2 https://zuul.opendev.org/t/openstack/build/83140c5c686b475a8c02fa7398d7e3e5 : SUCCESS in 21m 27s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/0868d5390483472c9b476c7778c99ff9 : SUCCESS in 32m 57s\n- cloudkitty-tempest-full-v2-storage-opensearch https://zuul.opendev.org/t/openstack/build/f733c592a0224324b73bfe9bf3759930 : SUCCESS in 44m 22s\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/d14b486e77c844f5aa74949ee8aa6f37 : SUCCESS in 18m 31s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/034156d07d4c428b964b28674ea1b2e5 : SUCCESS in 15m 09s\n- cloudkitty-tox-bandit https://zuul.opendev.org/t/openstack/build/99278389826d498c94a0d95e2dd71d92 : FAILURE in 1m 58s (non-voting)\n- cloudkitty-grenade-job https://zuul.opendev.org/t/openstack/build/406ac7a1dd244b1c9a80115b520325fe : FAILURE in 27m 32s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"75f9dd13bf59dfbdf64122cafa18c14cc6944334","revisions":{"75f9dd13bf59dfbdf64122cafa18c14cc6944334":{"kind":"REWORK","_number":1,"created":"2026-08-13 19:25:15.000000000","uploader":{"_account_id":35511,"name":"Dawud","email":"dawud@stackhpc.com","username":"dawudm"},"ref":"refs/changes/91/1000891/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty","ref":"refs/changes/91/1000891/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/91/1000891/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/91/1000891/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/91/1000891/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty refs/changes/91/1000891/1"}}},"commit":{"parents":[{"commit":"63f02806f9243dcd87b869234496bf8cfa500ef9","subject":"Fix OpenSearch mapping initialization","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/63f02806f9243dcd87b869234496bf8cfa500ef9"}]}],"author":{"name":"technowhizz","email":"7688823+technowhizz@users.noreply.github.com","date":"2026-08-06 00:03:27.000000000","tz":60},"committer":{"name":"Dawud","email":"dawud@stackhpc.com","date":"2026-08-13 19:19:39.000000000","tz":60},"subject":"Add storage backend migration command","message":"Add storage backend migration command\n\nAdd cloudkitty-storage-migrate for offline migrations between v2 storage\nbackends. Support automatic or explicit bounds, paginated copying,\ndestination overwrite, and post-copy verification.\n\nChange-Id: I11e8b9025ef2492be80e5c46e367153d02712330\nSigned-off-by: Dawud \u003cdawud@stackhpc.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/75f9dd13bf59dfbdf64122cafa18c14cc6944334"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/75f9dd13bf59dfbdf64122cafa18c14cc6944334"}]},"branch":"refs/heads/stable/2025.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
