)]}'
{"id":"openstack%2Fcloudkitty~696388","triplet_id":"openstack%2Fcloudkitty~master~If9aca985ebc04cc155d30aa49302413005a7ce97","project":"openstack/cloudkitty","branch":"master","hashtags":[],"change_id":"If9aca985ebc04cc155d30aa49302413005a7ce97","subject":"Switch import\u0027s behaviour to absolute imports for json module","status":"MERGED","created":"2019-11-27 18:44:19.000000000","updated":"2019-11-28 09:08:16.000000000","submitted":"2019-11-28 09:06:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696388-1574932013570-81677534","meta_rev_id":"131bc15c046d3ab832559c31003a8f75dcab9204","_number":696388,"virtual_id_number":696388,"owner":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-28 09:06:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"all":[{"value":2,"date":"2019-11-28 08:29:36.000000000","_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"all":[{"value":1,"date":"2019-11-28 08:29:36.000000000","_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 08:29:36.000000000","updated_by":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"reviewer":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"state":"REVIEWER"},{"updated":"2019-11-28 09:06:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5fd8c620b8773a3fe9c366ea627fda5f969a492d","author":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"date":"2019-11-27 18:44:19.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"71c4b6f2f5643b57dc69f8ded0ac5a57d7ba411e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-27 19:41:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/addcf5bf2f9a4e13ac24732a240af15c : SUCCESS in 5m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa3e387cf937488c95d6209cb44eef76 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/803a4a6d662441efbf06dd03bafaeee6 : SUCCESS in 4m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bc31d961481f4ce19150b93d723f2e1b : SUCCESS in 3m 29s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/db3f850a7c354916b0d91fd7fd578388 : SUCCESS in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb731aeb865441b7a1e63bb4d45725b7 : SUCCESS in 10m 43s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/1cfda1caec3c4341b0afb7bf0ca3081d : SUCCESS in 53m 02s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/b71806d4e77c41199e964110940b3959 : SUCCESS in 40m 26s (non-voting)\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/aee9299d515d4bcf878b91b6fa7cdc88 : SUCCESS in 35m 21s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/a4818e839a4d43059e21df3a34ddd937 : SUCCESS in 36m 36s\n- cloudkitty-tox-bandit https://zuul.opendev.org/t/openstack/build/e7fa7a1d6abe45f694b792219c347a4d : SUCCESS in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43dfdb6f4edb3189a232dea14c7631eda7ac46d0","author":{"_account_id":23060,"name":"Luka Peschke","email":"mail@lukapeschke.com","username":"lukapeschke"},"date":"2019-11-28 08:29:36.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks Joel, LGTM!","accounts_in_message":[],"_revision_number":1},{"id":"f32414f2fba29dfcfd395f999f01e698e8771d9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 08:29:47.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b0099a9c9b3f614818296e628a06218d8bf645b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:06:53.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/b3c219c28f4e49f982f6ddf70cb2d036 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7be04f78e3264a32aa9cf8f7f441e0bb : SUCCESS in 3m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e2ac6a9b69784f0282c15863df3f8a94 : SUCCESS in 6m 45s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/df5459a7586b4f6380f9cced2c2a97e6 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18401a64ae2f40fa8c202e60ee7b5d39 : SUCCESS in 6m 43s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/d9262316ee66428c9cd693a3d24bc4fd : SUCCESS in 36m 06s\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/8df2f63c453646a9a4e3c7bb172981e1 : SUCCESS in 35m 26s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/310a2527848b45cfbfde9c0c73221d0d : SUCCESS in 35m 31s","accounts_in_message":[],"_revision_number":1},{"id":"53eb2b0c4fef075864f6a563bd6852051a3d4e77","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:06:53.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"131bc15c046d3ab832559c31003a8f75dcab9204","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-28 09:08:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56daed61a3764400a464527fe61c67d0 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f0c3327e611be4da6c39c8964cf65beed5b04d9","revisions":{"0f0c3327e611be4da6c39c8964cf65beed5b04d9":{"kind":"REWORK","_number":1,"created":"2019-11-27 18:44:19.000000000","uploader":{"_account_id":31068,"name":"Joel Capitao","email":"jcapitao@redhat.com","username":"jcapiitao"},"ref":"refs/changes/88/696388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty","ref":"refs/changes/88/696388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/88/696388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/88/696388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/88/696388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty refs/changes/88/696388/1"}}},"commit":{"parents":[{"commit":"ee8b341f2d034e61a8a109f000c28678b3cc15df","subject":"Check for duplicates in \"groupby\" and \"metadata\" for each metric","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/ee8b341f2d034e61a8a109f000c28678b3cc15df"}]}],"author":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2019-11-27 18:32:47.000000000","tz":60},"committer":{"name":"Joel Capitao","email":"jcapitao@redhat.com","date":"2019-11-27 18:44:17.000000000","tz":60},"subject":"Switch import\u0027s behaviour to absolute imports for json module","message":"Switch import\u0027s behaviour to absolute imports for json module\n\nDownstream [1], when packaging cloudkitty, we hit the issue of\njson module importing itself, instead of standard library\u0027s\nversion.\nThis switch fixes the issue.\n\n[1] https://review.rdoproject.org/r/#/c/23860/\n\nChange-Id: If9aca985ebc04cc155d30aa49302413005a7ce97\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/0f0c3327e611be4da6c39c8964cf65beed5b04d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/0f0c3327e611be4da6c39c8964cf65beed5b04d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
