)]}'
{"id":"openstack%2Fcloudkitty~747057","triplet_id":"openstack%2Fcloudkitty~master~I4da9e8854a571058e48c2f51c1d340bc135cfe2b","project":"openstack/cloudkitty","branch":"master","topic":"replace-use-tz-util","hashtags":[],"change_id":"I4da9e8854a571058e48c2f51c1d340bc135cfe2b","subject":"Replace tz.UTC with dateutil.tz.tzutc()","status":"MERGED","created":"2020-08-19 23:04:16.000000000","updated":"2020-08-21 16:30:35.000000000","submitted":"2020-08-21 16:27:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"747057-1598027280118-0b925b31","meta_rev_id":"af0a8760224f44775273e69a806203962df1ca7a","_number":747057,"virtual_id_number":747057,"owner":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-08-21 16:27:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-08-21 15:01:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-08-21 15:01:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Gate Blocker Fix / Urgent Change"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-21 15:01:52.000000000","updated_by":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"reviewer":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"state":"REVIEWER"},{"updated":"2020-08-21 16:27:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"92a9ab0a8fb969500570491929279c9e790708ef","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-08-19 23:04:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"510724ebb588aefdbe578d5510c002dfd5e95d19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-19 23:47:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/15a9f3cadd4b489095a860d4260bea5a : SUCCESS in 5m 37s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/286d28ae0500445b928a199405218f3f : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e98a329564d4407c902f3a890677b058 : SUCCESS in 4m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7df56c9f089241ba82a3740e657d3871 : SUCCESS in 4m 58s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/4cc77992a59f4dfa8cc467bef5519608 : SUCCESS in 9m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8e9fb6df523b4553999226844e2aad38 : FAILURE in 4m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbaf2a0837d44106b45bb7122c6daa54 : SUCCESS in 5m 36s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/25d612f512634ed4a784b8f69eea2e9e : SUCCESS in 27m 05s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/ec16343ff4934e6ea228df4f0e0b4b2d : SUCCESS in 28m 41s (non-voting)\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/926d893bfc564d038a091a9a36bc7e21 : SUCCESS in 33m 15s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/23c0a82b4deb4977b4d0b7fb37f0ae9d : SUCCESS in 35m 43s\n- cloudkitty-tox-bandit https://zuul.opendev.org/t/openstack/build/ec12dbe26d234d8ea7546fac3e30182b : SUCCESS in 5m 02s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f19db1b82a742b7e692d08e9b5415e6a22b6d067","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 08:28:13.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2c5524e1f6b4a60bc0ff9f2e356c36b6d66e640","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 09:14:41.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/83a88fd7aeb44e7585720bae9f9c4b6f : SUCCESS in 7m 05s\n- requirements-check https://zuul.opendev.org/t/openstack/build/4c8b46a377924b53b97a60ff427dd8d3 : FAILURE in 2m 35s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee1d1e6df2784fe9a85db51bef3f6616 : SUCCESS in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a3afede6e1324ffe8fb8803f5ce824ac : SUCCESS in 4m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c67815d7a0934e52bf1bee2a5bc858e4 : SUCCESS in 4m 42s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/2c1705721d9c42c0962f937d712a11bb : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30ffc8d08c004855b9bdf2c15b694e2a : FAILURE in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d5420f789fdf47e490a0782258627368 : SUCCESS in 5m 52s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/41554c8e7a2b48d5ba8f9c0d5ee50af4 : SUCCESS in 33m 32s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/567f3cc653904eab89b0ead9fd2c03d5 : SUCCESS in 37m 19s (non-voting)\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/830255ce42c94cd5bcb85677f36eae9b : SUCCESS in 36m 25s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/153a49e8ee3e43b1b5ffd40412dd4feb : SUCCESS in 38m 11s\n- cloudkitty-tox-bandit https://zuul.opendev.org/t/openstack/build/b09068b9077f40a9a225092efb7f16ef : SUCCESS in 4m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4c09f5a7d535bbceca6a42af56e388ee66b0cf27","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 09:42:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5f86062e748cfff556c2d52b400c37a5b4b89b2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 10:21:56.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0f38e083166b486994f1b71c7ea21843 : SUCCESS in 5m 16s\n- requirements-check https://zuul.opendev.org/t/openstack/build/73684e7e41d944a29e08e0a79254c7b9 : SUCCESS in 3m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/074dad26a7b6424b99b91467988e8863 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a384247b854ca692b95e1e4fe195f0 : SUCCESS in 4m 40s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cf530f047769469ca23382e02ef8c809 : SUCCESS in 5m 31s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/0e80d6b665b34b2db69a9725effc6ae8 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/190d83b40c054352b4aa88837e732f52 : FAILURE in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90ed91e673b54069833726cedc407702 : SUCCESS in 8m 20s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/03bd5c66b73d4d078987134389987b88 : SUCCESS in 36m 06s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/265f23d2df374cd88c2c713d6998dabf : SUCCESS in 34m 28s (non-voting)\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/bf7a59374aea49c29e90ed8267d33eaf : SUCCESS in 34m 34s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/c3169cfab7914a6abf46bf3b2b84bd15 : SUCCESS in 34m 32s\n- cloudkitty-tox-bandit https://zuul.opendev.org/t/openstack/build/cc376cb5d0e34282942e7ef6a9faacc0 : SUCCESS in 4m 12s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"54ede0132dc7cb063c2aecc2d81d86f5507a3da4","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 12:55:19.000000000","message":"Patch Set 3: Code-Review+2\n\n@Rafael: I updated the patch slightly, please review","accounts_in_message":[],"_revision_number":3},{"id":"7789b34733ac820b287f0b98931469f91829758f","author":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"date":"2020-08-21 13:06:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"dda228129a0be3838932e6fd094d043ab12fe359","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 14:00:09.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"ec9ae01450d307d5671817ba204b13d9837d4c28","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 14:52:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6369c42ee7ef49518d452fec2b6a888e : SUCCESS in 5m 20s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c343226f76b94a4e9699681651bca9af : SUCCESS in 4m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bbdd2616e4694e77aea6dcc6feea314b : SUCCESS in 5m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d5dcece7798409a9ea2fa8142594e42 : SUCCESS in 5m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/471b90d054cd4c6daeb10a80bd5bb7bc : SUCCESS in 4m 44s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/b001df7827fb4dcf95fc0a351bfbace4 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7600d68a8e4b4d219f676470997882f1 : FAILURE in 6m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0146daa72eb9423ca853c6baeebe3fa5 : SUCCESS in 4m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ba526e339c94f88a901a1ed1a3faea0 : SUCCESS in 5m 00s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/75cc6e831e6e485f927ba732f8c7cc0e : SUCCESS in 36m 13s\n- cloudkitty-tempest-full-v2-storage-elasticsearch https://zuul.opendev.org/t/openstack/build/6cc393b24dc64eef8c600998a6c78f18 : SUCCESS in 40m 16s (non-voting)\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/6031bb47670240758fe4009af9c1e424 : SUCCESS in 44m 36s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/8bb7aa56a32044889ee0e284c538803e : SUCCESS in 41m 14s\n- cloudkitty-tox-bandit https://zuul.opendev.org/t/openstack/build/8de26180e3f9471d9ab7a0223401c988 : SUCCESS in 4m 45s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"35f19654d57defd3f154ced2b0b9fee844031648","author":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"date":"2020-08-21 15:01:52.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nAdded a release note, particularly for backports.","accounts_in_message":[],"_revision_number":4},{"id":"6b4089a7ac8725a570af4d21513b1f05b7989335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 15:02:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"02fe42e27b2bfcf49f7f8e3db5d6e986f8be8475","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 16:27:59.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/037bd007b6a345e3be5d5dbd0fb5dad3 : SUCCESS in 3m 55s\n- requirements-check https://zuul.opendev.org/t/openstack/build/dec2bda9df1a405a8a91db045d7f3950 : SUCCESS in 2m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/075afde54c94441daf72792bb52c6acd : SUCCESS in 4m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5899e5de3fe14555a4f18745437ba39a : SUCCESS in 6m 13s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/ab0b280716564a3cbf31b7661f481087 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e7478c6b1e4251a3b410b9974a03b1 : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82aa930c69e34f43b499f0331f4bc98a : SUCCESS in 3m 37s\n- cloudkitty-tempest-full-v2-storage-influxdb https://zuul.opendev.org/t/openstack/build/07b4c5834d454e5f89dbc9e7e484643f : SUCCESS in 35m 55s\n- cloudkitty-tempest-full-v1-storage-sqlalchemy https://zuul.opendev.org/t/openstack/build/85757bf33f564e089cacf1abf5cc1628 : SUCCESS in 44m 23s\n- cloudkitty-tempest-full-ipv6-only https://zuul.opendev.org/t/openstack/build/940cb7a76b0a45aeabd1e52390a59db6 : SUCCESS in 1h 19m 55s","accounts_in_message":[],"_revision_number":4},{"id":"6ff30567d06533feba15731134bd1af5b0df01bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 16:28:00.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"69ec1aaacade6bffe77404b4c9f03a789cb7a642","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-21 16:30:35.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14825a61069747e4b1b24170c8bb1c99 : SUCCESS in 1m 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7d5b41db6be74e5db9435896c8fa1d7d : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"b7abb9bb698d4d391da7f69ca983164f51d9a567","revisions":{"36f5e02dbaf767f17f14f22fa751a6b5409668f5":{"kind":"REWORK","_number":1,"created":"2020-08-19 23:04:16.000000000","uploader":{"_account_id":28356,"name":"Rafael Weingartner","email":"rafael@apache.org","username":"rafaelweingartner"},"ref":"refs/changes/57/747057/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty","ref":"refs/changes/57/747057/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/1"}}},"commit":{"parents":[{"commit":"8877fa1a37d8aafb3a7d238b002826e43439a1c9","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/8877fa1a37d8aafb3a7d238b002826e43439a1c9"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-19 23:03:51.000000000","tz":-180},"committer":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-19 23:03:51.000000000","tz":-180},"subject":"Replace tz.UTC with dateutil.tz.tzutc()","message":"Replace tz.UTC with dateutil.tz.tzutc()\n\nAs discussed in https://review.opendev.org/#/c/742477/,\nthis patch replaces the use of `tx.UTC` with `dateutil.tz.tzutc()`\n\nChange-Id: I4da9e8854a571058e48c2f51c1d340bc135cfe2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/36f5e02dbaf767f17f14f22fa751a6b5409668f5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/36f5e02dbaf767f17f14f22fa751a6b5409668f5"}]},"branch":"refs/heads/master"},"c45ba427c37279db816ca01489d5c42971cec178":{"kind":"REWORK","_number":2,"created":"2020-08-21 08:28:13.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/747057/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty","ref":"refs/changes/57/747057/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/2"}}},"commit":{"parents":[{"commit":"8877fa1a37d8aafb3a7d238b002826e43439a1c9","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/8877fa1a37d8aafb3a7d238b002826e43439a1c9"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-19 23:03:51.000000000","tz":-180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 08:19:26.000000000","tz":120},"subject":"Replace tz.UTC with dateutil.tz.tzutc()","message":"Replace tz.UTC with dateutil.tz.tzutc()\n\nAs discussed in https://review.opendev.org/#/c/742477/, this patch\nreplaces the use of `tz.UTC` with `dateutil.tz.tzutc()`\n\nAlso add python-dateutil to requirements.txt, using the same minimum\nversion as several other OpenStack projects, including nova.\n\nChange-Id: I4da9e8854a571058e48c2f51c1d340bc135cfe2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/c45ba427c37279db816ca01489d5c42971cec178"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/c45ba427c37279db816ca01489d5c42971cec178"}]},"branch":"refs/heads/master"},"3f60749dba91a73e65b53cfeecea8e3278dde943":{"kind":"REWORK","_number":3,"created":"2020-08-21 09:42:50.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/747057/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty","ref":"refs/changes/57/747057/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/3"}}},"commit":{"parents":[{"commit":"8877fa1a37d8aafb3a7d238b002826e43439a1c9","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/8877fa1a37d8aafb3a7d238b002826e43439a1c9"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-19 23:03:51.000000000","tz":-180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 09:42:42.000000000","tz":120},"subject":"Replace tz.UTC with dateutil.tz.tzutc()","message":"Replace tz.UTC with dateutil.tz.tzutc()\n\nAs discussed in https://review.opendev.org/#/c/742477/, this patch\nreplaces the use of `tz.UTC` with `dateutil.tz.tzutc()`\n\nAlso add python-dateutil to requirements.txt, using the same minimum\nversion as several other OpenStack projects, including nova.\n\nChange-Id: I4da9e8854a571058e48c2f51c1d340bc135cfe2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/3f60749dba91a73e65b53cfeecea8e3278dde943"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/3f60749dba91a73e65b53cfeecea8e3278dde943"}]},"branch":"refs/heads/master"},"b7abb9bb698d4d391da7f69ca983164f51d9a567":{"kind":"REWORK","_number":4,"created":"2020-08-21 14:00:09.000000000","uploader":{"_account_id":15197,"name":"Pierre Riteau","email":"pierre@stackhpc.com","username":"priteau","status":"StackHPC"},"ref":"refs/changes/57/747057/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cloudkitty","ref":"refs/changes/57/747057/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cloudkitty refs/changes/57/747057/4"}}},"commit":{"parents":[{"commit":"8877fa1a37d8aafb3a7d238b002826e43439a1c9","subject":"Merge \"Use unittest.mock instead of third party mock\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/8877fa1a37d8aafb3a7d238b002826e43439a1c9"}]}],"author":{"name":"Rafael Weingärtner","email":"rafael@apache.org","date":"2020-08-19 23:03:51.000000000","tz":-180},"committer":{"name":"Pierre Riteau","email":"pierre@stackhpc.com","date":"2020-08-21 13:58:38.000000000","tz":120},"subject":"Replace tz.UTC with dateutil.tz.tzutc()","message":"Replace tz.UTC with dateutil.tz.tzutc()\n\nAs discussed in https://review.opendev.org/#/c/742477/, this patch\nreplaces the use of `tz.UTC` with `dateutil.tz.tzutc()`\n\nAlso add python-dateutil to requirements.txt, using the same minimum\nversion as several other OpenStack projects, including nova.\n\nChange-Id: I4da9e8854a571058e48c2f51c1d340bc135cfe2b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/b7abb9bb698d4d391da7f69ca983164f51d9a567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cloudkitty/commit/b7abb9bb698d4d391da7f69ca983164f51d9a567"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
