)]}'
{"id":"openstack%2Fcodegenerator~936460","triplet_id":"openstack%2Fcodegenerator~master~I3f583deb9b3c310a9b05371e583ba570f80bbb77","project":"openstack/codegenerator","branch":"master","hashtags":[],"change_id":"I3f583deb9b3c310a9b05371e583ba570f80bbb77","subject":"Only compile generated rust code","status":"MERGED","created":"2024-11-28 13:56:13.000000000","updated":"2024-11-28 16:39:45.000000000","submitted":"2024-11-28 16:38:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"936460","meta_rev_id":"8a38f58b006285467d745ebc18a470a8934c6c2a","_number":936460,"virtual_id_number":936460,"owner":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-28 16:38:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":2,"date":"2024-11-28 15:08:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"all":[{"value":1,"date":"2024-11-28 15:08:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-28 15:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"adbb747fa4eb88c9f940b6c3754123d4192d6071","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-28 13:56:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3efdea4e3d2c27ac6b018c57a8947bdcb738d960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 15:07:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e2b68ee7c5f94c3da438a7c814b18e48\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f1cb74ea6d4cf99a75d5b5229d0477 : SUCCESS in 2m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7872e74d608549c0a6f3c8c5b531e43a : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73f617a2ccae43f5aee0a8e731b19831 : SUCCESS in 3m 17s\n- codegenerator-openapi-baremetal-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/57c5f366f881413eb84de9daec3fbc20 : SUCCESS in 3m 18s\n- codegenerator-openapi-block-storage-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/0daf0e3918c9431a8eba1bac20cbdb0b : SUCCESS in 7m 13s\n- codegenerator-openapi-compute-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/d49e6f467deb4170ac110b104defbbd5 : SUCCESS in 5m 52s (non-voting)\n- codegenerator-openapi-dns-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/13dd60e04fc14eb6abcf281558a0bb65 : SUCCESS in 3m 14s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/1340edd1f3864e32bad624d131824c23 : SUCCESS in 4m 59s\n- codegenerator-openapi-image-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c431037bfcf54bef8028a6e64703d057 : SUCCESS in 4m 14s\n- codegenerator-openapi-load-balancing-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/731e5c4be8b548769dc758316b82c29c : SUCCESS in 4m 56s\n- codegenerator-openapi-network-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e6408f042d4d493ebad57e6881943dde : SUCCESS in 6m 35s\n- codegenerator-openapi-object-store-static https://zuul.opendev.org/t/openstack/build/3e89b03dc30d48f4890ca1a507f1be05 : SUCCESS in 1m 17s\n- codegenerator-openapi-placement-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/40371aa6eadb461290ccf88a1a46569b : SUCCESS in 6m 16s\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/e2760e4b3630447bbec8bf4ebc9b6f24 : SUCCESS in 5m 27s\n- codegenerator-tox-publish-openapi-specs https://zuul.opendev.org/t/openstack/build/d5c81c6adb44456c837dc43c6bb49c49 : SUCCESS in 3m 06s\n- codegenerator-rust-all https://zuul.opendev.org/t/openstack/build/70009d5b38784acea08a0ef1b884286d : SUCCESS in 49m 34s","accounts_in_message":[],"_revision_number":1},{"id":"3aaec1bd89536e2bc119bd6ecfff05036f1a3e0b","author":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"date":"2024-11-28 15:08:04.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"a243e15480fdd168d1f5da8e2ca63953f360a766","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 15:08:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b8372f00c136ecd1e4075e5c47573f3908ada405","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 16:38:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/403520876cf44b7fb10c6ba3f77c4919\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97447bb60d2a41be966d5c2e4dc2d567 : SUCCESS in 2m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f1bb0250c1aa4266b7242ee1122cdbbf : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b4131785afb46a298297c464e19b759 : SUCCESS in 3m 45s\n- codegenerator-openapi-baremetal-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/676563d4521f41d998a48809862616fd : SUCCESS in 4m 49s\n- codegenerator-openapi-block-storage-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/eb974e0321524681b1c09966de70661a : SUCCESS in 6m 33s\n- codegenerator-openapi-compute-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/a4c84e460f3f427cb9a24fdafc8ab061 : SUCCESS in 8m 43s (non-voting)\n- codegenerator-openapi-dns-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/432cf137db674a59886d0eaa883ec5c4 : SUCCESS in 4m 48s\n- codegenerator-openapi-identity-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/6925a7b7dd1b4bf9a57a51af76b26156 : SUCCESS in 6m 34s\n- codegenerator-openapi-image-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/175bf90d50b04f8cb0df9ce35d8a1eb4 : SUCCESS in 5m 09s\n- codegenerator-openapi-load-balancing-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/42d8d6232778436a96d991ead03b4952 : SUCCESS in 4m 25s\n- codegenerator-openapi-network-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/5c265f8b0fbd47688688ed9b03db04f7 : SUCCESS in 9m 06s\n- codegenerator-openapi-object-store-static https://zuul.opendev.org/t/openstack/build/d1dc479a9f3c4d1197f74095a66f545d : SUCCESS in 2m 21s\n- codegenerator-openapi-placement-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/ab55df813d804ab49141b00f4421b5eb : SUCCESS in 3m 05s\n- codegenerator-openapi-shared-file-system-tips-with-api-ref https://zuul.opendev.org/t/openstack/build/c56d03b81a024e5fbcce8de014a38dcb : SUCCESS in 7m 51s\n- codegenerator-tox-publish-openapi-specs https://zuul.opendev.org/t/openstack/build/43aecfd2c398455eb7dce8dd414ad2aa : SUCCESS in 3m 09s\n- codegenerator-rust-all https://zuul.opendev.org/t/openstack/build/d5672dea78c74facb8576ef74140960d : SUCCESS in 49m 28s","accounts_in_message":[],"_revision_number":1},{"id":"122b7e858df2c17ae53bf2a8380e6ed345cbaf2f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 16:38:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8a38f58b006285467d745ebc18a470a8934c6c2a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 16:39:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b299de692b744583b65a4c4cc0629a3a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/12c29fc6f0254e9f874f2249cf4112f7 : SUCCESS in 39s\n- codegenerator-propose-rust-openstack-change https://zuul.opendev.org/t/openstack/build/a18bb708e3ed4355b8f0026b55005bfa : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b3465736170f0585e64012f349bebeed13e706b8","revisions":{"b3465736170f0585e64012f349bebeed13e706b8":{"kind":"REWORK","_number":1,"created":"2024-11-28 13:56:13.000000000","uploader":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"},"ref":"refs/changes/60/936460/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/codegenerator","ref":"refs/changes/60/936460/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/codegenerator refs/changes/60/936460/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/codegenerator refs/changes/60/936460/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/codegenerator refs/changes/60/936460/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/codegenerator refs/changes/60/936460/1"}}},"commit":{"parents":[{"commit":"cadf4f865257faded417140df0da5d353b4d56f2","subject":"Add octavia query parameters","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/codegenerator/commit/cadf4f865257faded417140df0da5d353b4d56f2"}]}],"author":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-28 13:53:35.000000000","tz":60},"committer":{"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","date":"2024-11-28 13:53:35.000000000","tz":60},"subject":"Only compile generated rust code","message":"Only compile generated rust code\n\nWe should not try to compile the code that is not being generated. This\nhelps preventing breaking changes but at the same time slows down\npropagation of the real API changes. Downstream tools (i.e.\nopenstack_tui) need to be adapted in the PR propagating the sdk change\nuntil we start generating tui code as well.\n\nChange-Id: I3f583deb9b3c310a9b05371e583ba570f80bbb77\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/codegenerator/commit/b3465736170f0585e64012f349bebeed13e706b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/codegenerator/commit/b3465736170f0585e64012f349bebeed13e706b8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":27900,"name":"Artem Goncharov","email":"artem.goncharov@gmail.com","username":"gtema"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
