)]}'
{"id":"openstack%2Fcongress-tempest-plugin~721740","triplet_id":"openstack%2Fcongress-tempest-plugin~master~I74824174861c394e75b8c733ae9bbc3c9b410268","project":"openstack/congress-tempest-plugin","branch":"master","topic":"retire-congress","hashtags":[],"change_id":"I74824174861c394e75b8c733ae9bbc3c9b410268","subject":"Retire the Congress Tempest Plugin project","status":"MERGED","created":"2020-04-21 21:50:16.000000000","updated":"2020-05-28 17:12:31.000000000","submitted":"2020-05-28 17:12:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"721740-1590685951783-c6985dc5","meta_rev_id":"4e280dba4680e8b98b0546f8cb3d2b9dc1afd38d","_number":721740,"virtual_id_number":721740,"owner":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-05-28 17:12:31.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"recommended":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"all":[{"value":1,"date":"2020-05-28 16:17:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2020-05-28 16:28:27.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"all":[{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2020-05-28 16:28:27.000000000","_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-28 16:17:12.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2020-05-28 17:12:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e090a95105f9d63e418b850b352605c9644ac542","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-04-21 21:50:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d0cca2a1f94f7876794d80e02e72c3b03a5c331","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-04-21 21:51:57.000000000","message":"Patch Set 1: Workflow-1\n\nMarking this -W until the Ussuri release.","accounts_in_message":[],"_revision_number":1},{"id":"6a8f4c07b355c773346856a6140ae79a73e4d6ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-21 21:56:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6fbf4ecd5e894f54a7c5c1145e9e76fa : FAILURE in 1m 59s\n- noop https://zuul.opendev.org/t/openstack/build/ca782cde4ebb4b20a2010beaa8aebd71 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"84a3e16e8eaa899c8bded811dc4251cff20a0431","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-04-22 06:01:48.000000000","message":"Patch Set 1: Code-Review-1\n\nNeeds adding of noop-jobs to project-config first - and then remove .zuul.yaml here as well","accounts_in_message":[],"_revision_number":1},{"id":"0e7781c872c56184a22abc9829b16dfae4b29237","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-27 22:12:47.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e35d1597aff86f5b3168238b171ce1c5f914b23b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-27 22:17:32.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/7be8eadd05a34695a13fc7f05f0d7c17 : FAILURE in 2m 19s","accounts_in_message":[],"_revision_number":2},{"id":"92e5e5c2b8c4538fff2ca2267c73dbe51a2e508e","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-28 15:45:55.000000000","message":"Patch Set 2:\n\nrecheck noop-jobs merged","accounts_in_message":[],"_revision_number":2},{"id":"ecc948baa59722510ef89f3eebf017cf4f153fd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 15:46:08.000000000","message":"Patch Set 2:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":2},{"id":"5107f30cd1daf3353bccace94d75eae17b82ad56","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-28 15:50:23.000000000","message":"Patch Set 2:\n\nrecheck merge conflict","accounts_in_message":[],"_revision_number":2},{"id":"3f3e1da1bfcca4b9e16ce6e9de68950f431af8bb","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-28 16:13:24.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"a5d31773af1f37f88f79dd9a9418057f837fb264","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-28 16:15:22.000000000","message":"Patch Set 3:\n\nneeds noop-job in project-config...","accounts_in_message":[],"_revision_number":3},{"id":"81714ae43e07fed95e5e6b75080712cfb8be7f63","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-28 16:15:26.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"bfc3a69d2f71a87f33859d25655bb5134044c6b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:16:23.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/80435597984144f6a84385333cfb04e6 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":4},{"id":"fed2455b0a655adf833b1cc413a93b2d05866ae3","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2020-05-28 16:17:12.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"c77469834ab64da3577e7056108f5e030e008a68","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2020-05-28 16:28:27.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\nAll open reviews have been closed.  Project is now retired.","accounts_in_message":[],"_revision_number":4},{"id":"6c39a858bb3a794659632ff9e523a245d2755c42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 16:28:45.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"36a7e18586e843112b3541c285540ae67dace031","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:12:31.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"9ed9beb1459b240f6bfa39c21d583004a94fd5d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-28 17:12:31.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/5ed9564693534cfdaa72e7a639607e12 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"3d3098c8938eec9fe427f3515c90787840d6c903","revisions":{"27a632117133e1fb6b3a716271fccc9eed90c143":{"kind":"REWORK","_number":1,"created":"2020-04-21 21:50:16.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/40/721740/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/congress-tempest-plugin","ref":"refs/changes/40/721740/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/1"}}},"commit":{"parents":[{"commit":"f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-04-21 21:48:51.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-04-21 21:50:15.000000000","tz":-240},"subject":"Retire the Congress Tempest Plugin project","message":"Retire the Congress Tempest Plugin project\n\nRecently the TC has worked on determining the criteria for when an\nOpenStack project should be retired.  When there was not a PTL nominee\nfor the Congress project, that triggered the TC to review the project\nhealth per [1], and the TC has determined [2] that development work on\nthe project has ceased.  This decision was announced in the\nopenstack-discuss mailing list in April 2020 [3].\n\nThis commit retires the repository per the process for governance\nremoval in the Victoria cycle as specified in the Mandatory Repository\nRetirement resolution [4] and detailed in the infra manual [5].\n\nShould interest in developing Congress as part of OpenStack revive,\nplease revert this commit to have the project rejoin the list of active\nprojects.\n\nThe community wishes to express our thanks and appreciation to all of\nthose who have contributed to the Congress project over the years.\n\n[1] https://governance.openstack.org/tc/reference/dropping-projects.html\n[2] http://eavesdrop.openstack.org/irclogs/%23openstack-tc/latest.log.html#t2020-04-20T15:36:59\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n[4] https://governance.openstack.org/tc/resolutions/20190711-mandatory-repository-retirement.html\n[5] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#retiring-a-project\n\nChange-Id: I74824174861c394e75b8c733ae9bbc3c9b410268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/27a632117133e1fb6b3a716271fccc9eed90c143"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/27a632117133e1fb6b3a716271fccc9eed90c143"}]},"branch":"refs/heads/master"},"2432c802b3067d9908c78e8e9362d448afa4ab5b":{"kind":"REWORK","_number":2,"created":"2020-05-27 22:12:47.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/40/721740/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/congress-tempest-plugin","ref":"refs/changes/40/721740/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/2"}}},"commit":{"parents":[{"commit":"f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-04-21 21:48:51.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-27 22:12:30.000000000","tz":-240},"subject":"Retire the Congress Tempest Plugin project","message":"Retire the Congress Tempest Plugin project\n\nRecently the TC has worked on determining the criteria for when an\nOpenStack project should be retired.  When there was not a PTL nominee\nfor the Congress project, that triggered the TC to review the project\nhealth per [1], and the TC has determined [2] that development work on\nthe project has ceased.  This decision was announced in the\nopenstack-discuss mailing list in April 2020 [3].\n\nThis commit retires the repository per the process for governance\nremoval in the Victoria cycle as specified in the Mandatory Repository\nRetirement resolution [4] and detailed in the infra manual [5].\n\nShould interest in developing Congress as part of OpenStack revive,\nplease revert this commit to have the project rejoin the list of active\nprojects.\n\nThe community wishes to express our thanks and appreciation to all of\nthose who have contributed to the Congress project over the years.\n\n[1] https://governance.openstack.org/tc/reference/dropping-projects.html\n[2] http://eavesdrop.openstack.org/irclogs/%23openstack-tc/latest.log.html#t2020-04-20T15:36:59\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n[4] https://governance.openstack.org/tc/resolutions/20190711-mandatory-repository-retirement.html\n[5] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#retiring-a-project\n\nDepends-On: https://review.opendev.org/721742\n\nChange-Id: I74824174861c394e75b8c733ae9bbc3c9b410268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/2432c802b3067d9908c78e8e9362d448afa4ab5b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/2432c802b3067d9908c78e8e9362d448afa4ab5b"}]},"branch":"refs/heads/master"},"3078297ffb85dbad58a4106bd71603cc7faf7394":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-05-28 16:13:24.000000000","uploader":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"ref":"refs/changes/40/721740/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/congress-tempest-plugin","ref":"refs/changes/40/721740/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/3"}}},"commit":{"parents":[{"commit":"f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-04-21 21:48:51.000000000","tz":-240},"committer":{"name":"Andreas Jaeger","email":"aj@suse.com","date":"2020-05-28 16:13:14.000000000","tz":120},"subject":"Retire the Congress Tempest Plugin project","message":"Retire the Congress Tempest Plugin project\n\nRecently the TC has worked on determining the criteria for when an\nOpenStack project should be retired.  When there was not a PTL nominee\nfor the Congress project, that triggered the TC to review the project\nhealth per [1], and the TC has determined [2] that development work on\nthe project has ceased.  This decision was announced in the\nopenstack-discuss mailing list in April 2020 [3].\n\nThis commit retires the repository per the process for governance\nremoval in the Victoria cycle as specified in the Mandatory Repository\nRetirement resolution [4] and detailed in the infra manual [5].\n\nShould interest in developing Congress as part of OpenStack revive,\nplease revert this commit to have the project rejoin the list of active\nprojects.\n\nThe community wishes to express our thanks and appreciation to all of\nthose who have contributed to the Congress project over the years.\n\n[1] https://governance.openstack.org/tc/reference/dropping-projects.html\n[2] http://eavesdrop.openstack.org/irclogs/%23openstack-tc/latest.log.html#t2020-04-20T15:36:59\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n[4] https://governance.openstack.org/tc/resolutions/20190711-mandatory-repository-retirement.html\n[5] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#retiring-a-project\n\nChange-Id: I74824174861c394e75b8c733ae9bbc3c9b410268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/3078297ffb85dbad58a4106bd71603cc7faf7394"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/3078297ffb85dbad58a4106bd71603cc7faf7394"}]},"branch":"refs/heads/master"},"3d3098c8938eec9fe427f3515c90787840d6c903":{"kind":"NO_CHANGE","_number":4,"created":"2020-05-28 16:15:26.000000000","uploader":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"ref":"refs/changes/40/721740/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/congress-tempest-plugin","ref":"refs/changes/40/721740/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/congress-tempest-plugin refs/changes/40/721740/4"}}},"commit":{"parents":[{"commit":"f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/f924c6cd9876b2a2e09a4d5ddddf7b489ac40a6e"}]}],"author":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-04-21 21:48:51.000000000","tz":-240},"committer":{"name":"Nate Johnston","email":"iam@natejohnston.info","date":"2020-05-28 16:15:14.000000000","tz":-240},"subject":"Retire the Congress Tempest Plugin project","message":"Retire the Congress Tempest Plugin project\n\nRecently the TC has worked on determining the criteria for when an\nOpenStack project should be retired.  When there was not a PTL nominee\nfor the Congress project, that triggered the TC to review the project\nhealth per [1], and the TC has determined [2] that development work on\nthe project has ceased.  This decision was announced in the\nopenstack-discuss mailing list in April 2020 [3].\n\nThis commit retires the repository per the process for governance\nremoval in the Victoria cycle as specified in the Mandatory Repository\nRetirement resolution [4] and detailed in the infra manual [5].\n\nShould interest in developing Congress as part of OpenStack revive,\nplease revert this commit to have the project rejoin the list of active\nprojects.\n\nThe community wishes to express our thanks and appreciation to all of\nthose who have contributed to the Congress project over the years.\n\n[1] https://governance.openstack.org/tc/reference/dropping-projects.html\n[2] http://eavesdrop.openstack.org/irclogs/%23openstack-tc/latest.log.html#t2020-04-20T15:36:59\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2020-April/014292.html\n[4] https://governance.openstack.org/tc/resolutions/20190711-mandatory-repository-retirement.html\n[5] https://docs.opendev.org/opendev/infra-manual/latest/drivers.html#retiring-a-project\n\nChange-Id: I74824174861c394e75b8c733ae9bbc3c9b410268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/3d3098c8938eec9fe427f3515c90787840d6c903"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/congress-tempest-plugin/commit/3d3098c8938eec9fe427f3515c90787840d6c903"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
