)]}'
{"id":"openstack%2Fcontributor-guide~555725","triplet_id":"openstack%2Fcontributor-guide~master~I725920daecf28675c656ee8aca0b7851321c741d","project":"openstack/contributor-guide","branch":"master","topic":"tracking-intro","hashtags":[],"change_id":"I725920daecf28675c656ee8aca0b7851321c741d","subject":"Added brief overview of the whole section.","status":"MERGED","created":"2018-03-23 10:10:03.000000000","updated":"2018-03-29 17:27:36.000000000","submitted":"2018-03-29 17:27:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"555725-1522344456627-463e99ac","meta_rev_id":"94b731152d7d762508a41510c8737919fdf25aba","_number":555725,"virtual_id_number":555725,"owner":{"_account_id":22534,"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","username":"visnyei"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":0,"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":2,"date":"2018-03-29 17:27:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"recommended":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"all":[{"value":1,"date":"2018-03-28 23:06:37.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":2,"date":"2018-03-23 13:23:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2018-03-28 23:52:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"all":[{"value":0,"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"value":1,"date":"2018-03-29 17:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-03-28 23:06:37.000000000","updated_by":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"reviewer":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"state":"REVIEWER"},{"updated":"2018-03-28 23:52:34.000000000","updated_by":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"reviewer":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"state":"REVIEWER"},{"updated":"2018-03-29 17:23:59.000000000","updated_by":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"reviewer":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"state":"REVIEWER"},{"updated":"2018-03-29 17:27:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"2a5c5a6093f4d853cf2601cf5647860f64249bd7","author":{"_account_id":22534,"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","username":"visnyei"},"date":"2018-03-23 10:10:03.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d966324a2b180a32a8f889585d013c495ddc7346","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-23 10:13:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/25/555725/1/check/build-openstack-sphinx-docs/a44c78b/html/ : SUCCESS in 2m 36s\n- openstack-tox-pep8 http://logs.openstack.org/25/555725/1/check/openstack-tox-pep8/c103ff9/ : SUCCESS in 2m 11s","accounts_in_message":[],"_revision_number":1},{"id":"97d840c047d7f8c5d094489a098f28daa915e244","author":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"date":"2018-03-23 13:23:49.000000000","message":"Patch Set 1: Code-Review+2\n\nlgtm, thx","accounts_in_message":[],"_revision_number":1},{"id":"a15b09652fa5ecef4e15bb4ce6ce55e76ef5bc66","author":{"_account_id":14482,"name":"Ian Y. Choi","email":"ianyrchoi@gmail.com","username":"ianychoi"},"date":"2018-03-28 23:06:37.000000000","message":"Patch Set 1: Code-Review+1\n\nLooks good :)","accounts_in_message":[],"_revision_number":1},{"id":"252f5496672f4f9ba3a6047a2a8d217c4e4a1817","author":{"_account_id":21486,"name":"wangqi","email":"wang.qi@99cloud.net","username":"wangqi"},"date":"2018-03-28 23:52:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"da553f72ce523237026eac104dba5aa388177565","author":{"_account_id":17765,"name":"Frank Kloeker","email":"f.kloeker@telekom.de","username":"eumel8"},"date":"2018-03-29 17:23:59.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"92a6e1bc7bee13bdcacbc4be8f9e2793b1b04296","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-29 17:24:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6dad85664f2264fafd81f6800fb7cdde0fc48e5a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-29 17:27:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- build-openstack-sphinx-docs http://logs.openstack.org/25/555725/1/gate/build-openstack-sphinx-docs/2409751/html/ : SUCCESS in 2m 25s\n- openstack-tox-pep8 http://logs.openstack.org/25/555725/1/gate/openstack-tox-pep8/f13385c/ : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":1},{"id":"9bcf082c9f760a774e3f6d1f1bf5619ef572f041","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-03-29 17:27:36.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"74b93b9f77be6737977acfd120cd20b72a373bc0","revisions":{"74b93b9f77be6737977acfd120cd20b72a373bc0":{"kind":"REWORK","_number":1,"created":"2018-03-23 10:10:03.000000000","uploader":{"_account_id":22534,"name":"Andrea Visnyei","email":"andrea.visnyei@nokia.com","username":"visnyei"},"ref":"refs/changes/25/555725/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/contributor-guide","ref":"refs/changes/25/555725/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/25/555725/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/25/555725/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/25/555725/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/contributor-guide refs/changes/25/555725/1"}}},"commit":{"parents":[{"commit":"1e51e48349f3ed86a76ca57207d70ec9f27f79e1","subject":"Merge \"Moving the Communication slides to a new section\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/1e51e48349f3ed86a76ca57207d70ec9f27f79e1"}]}],"author":{"name":"visnyei","email":"andrea.visnyei@nokia.com","date":"2018-03-23 10:09:01.000000000","tz":60},"committer":{"name":"visnyei","email":"andrea.visnyei@nokia.com","date":"2018-03-23 10:09:01.000000000","tz":60},"subject":"Added brief overview of the whole section.","message":"Added brief overview of the whole section.\n\nStory: 2001607\n\nTask: 6603\n\nChange-Id: I725920daecf28675c656ee8aca0b7851321c741d\nSigned-off-by: visnyei \u003candrea.visnyei@nokia.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/74b93b9f77be6737977acfd120cd20b72a373bc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/74b93b9f77be6737977acfd120cd20b72a373bc0"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"1e51e48349f3ed86a76ca57207d70ec9f27f79e1","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
