)]}'
{"id":"openstack%2Fcontributor-guide~634333","triplet_id":"openstack%2Fcontributor-guide~master~I2e00505621ba4fec99e98bc1bfb1e37d2468d96d","project":"openstack/contributor-guide","branch":"master","topic":"change-series-tutorial","hashtags":[],"change_id":"I2e00505621ba4fec99e98bc1bfb1e37d2468d96d","subject":"Tutorial: changes in a series","status":"MERGED","created":"2019-01-31 22:28:18.000000000","updated":"2019-02-11 21:11:58.000000000","submitted":"2019-02-11 21:11:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"634333-1549919518878-f954b26b","meta_rev_id":"17078da28c3c7eb761a0e27df258846f9311d017","_number":634333,"virtual_id_number":634333,"owner":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":2,"date":"2019-02-11 21:11:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":2,"date":"2019-02-11 21:04:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-02-07 20:49:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"value":0,"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"value":0,"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"value":1,"date":"2019-02-11 21:04:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"value":0,"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-01-31 22:33:15.000000000","updated_by":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"reviewer":{"_account_id":2472,"name":"Doug Hellmann","email":"dhellmann@redhat.com","username":"doug-hellmann"},"state":"REVIEWER"},{"updated":"2019-01-31 22:33:17.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-01-31 22:33:26.000000000","updated_by":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"reviewer":{"_account_id":4146,"name":"Clark Boylan","email":"cboylan@sapwetik.org","username":"cboylan"},"state":"REVIEWER"},{"updated":"2019-01-31 22:33:38.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2019-01-31 22:33:59.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-01-31 22:44:59.000000000","updated_by":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"reviewer":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"state":"REVIEWER"},{"updated":"2019-02-01 15:53:05.000000000","updated_by":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"reviewer":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"state":"REVIEWER"},{"updated":"2019-02-07 20:49:33.000000000","updated_by":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"reviewer":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"state":"REVIEWER"},{"updated":"2019-02-11 21:04:44.000000000","updated_by":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"reviewer":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"state":"REVIEWER"},{"updated":"2019-02-11 21:11:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"80e8becf0456fcb9ab63ef80fc6dd2bb86ade53b","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-01-31 22:28:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1d205886aacf9511ce5b8f9ea3bd1b738f0750e4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-01-31 22:33:02.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs http://logs.openstack.org/33/634333/1/check/openstack-tox-docs/521ff1a/ : FAILURE in 3m 12s\n- openstack-tox-pep8 http://logs.openstack.org/33/634333/1/check/openstack-tox-pep8/bcefd8e/ : FAILURE in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"40551c3b76f525cd5083e23c2bb5636269732e03","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-01-31 22:42:43.000000000","message":"Patch Set 1:\n\nDocs build is failing for some now-fatal warnings. Will stuff another patch under this one to fix those up.\n\nThe pep8 failures are legit; will fix when I restack.","accounts_in_message":[],"_revision_number":1},{"id":"0b9d4f9b1ee62a5df8748d24a266815290652095","author":{"_account_id":10608,"name":"Matthew Edmonds","email":"edmondsw@us.ibm.com","username":"edmondsw"},"date":"2019-01-31 22:44:59.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"48171e6bee14427dd052f6ca9423f382800d74c8","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-01-31 23:20:43.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"608f755ed56b0fb309c89ca475d5437f7944bd4a","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-01 15:46:25.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a4cb2ab67445a9a23bc50b486fddd8584e119dff","author":{"_account_id":6547,"name":"Andreas Jaeger","email":"jaegerandi@gmail.com","username":"jaegerandi"},"date":"2019-02-01 15:53:05.000000000","message":"Patch Set 2:\n\nthis should build without the other one,  please rebase on master and see the latest change that merged.","accounts_in_message":[],"_revision_number":2},{"id":"ff1e5958d67984a0f694dda53e095a4ebf72cb6a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-01 15:55:08.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/634333/2/check/openstack-tox-docs/eefe5e8/html/ : SUCCESS in 3m 32s\n- openstack-tox-pep8 http://logs.openstack.org/33/634333/2/check/openstack-tox-pep8/6e2216e/ : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":2},{"id":"b94bbf23289de16d9e6126802029aefd093981e0","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-01 16:02:57.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"37efdc841d809dbde696d5f51b4217e4f2924ac1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-01 16:10:48.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/634333/3/check/openstack-tox-docs/cf8a377/html/ : SUCCESS in 5m 00s\n- openstack-tox-pep8 http://logs.openstack.org/33/634333/3/check/openstack-tox-pep8/21a3334/ : SUCCESS in 3m 14s","accounts_in_message":[],"_revision_number":3},{"id":"9b12df0bfac24ebbddacde6b2d3b4adc526482bc","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-01 19:06:07.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"52dfa2319d70af99d8a40b25724f1b65e3245344","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-01 19:06:35.000000000","message":"Patch Set 4:\n\n\u003e this should build without the other one,  please rebase on master and see the latest change that merged.\n\nDone.","accounts_in_message":[],"_revision_number":4},{"id":"6e37934d1c9bec2be007204ed63558b5ef623062","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-01 19:12:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/634333/4/check/openstack-tox-docs/8e6d50c/html/ : SUCCESS in 4m 53s\n- openstack-tox-pep8 http://logs.openstack.org/33/634333/4/check/openstack-tox-pep8/a70c816/ : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":4},{"id":"c5d59185b842d016ddb5f87703fd16a655cbaef5","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2019-02-05 18:00:22.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)\n\nHi Eric,\n\nThank you for adding this information to the Contributor Guide. I think it\u0027s a very useful piece as working with patch chains has its challenges.\n\nI added a comment in line regarding the example itself which guides through the process as I think it would be better to make sure people play with the sandbox repository if they are not sure what they are doing.\n\nThanks,\nIldikó","accounts_in_message":[],"_revision_number":4},{"id":"996172a86322e1d95df4e65ff382287ff531368f","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-06 22:32:15.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e37a4cec7d60862a34b3fee80ed3f8a9bade420e","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-06 22:32:40.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d9a4a38ecda028504c7a3735eb71f07348242a1b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-06 22:41:09.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/634333/5/check/openstack-tox-docs/67eae08/html/ : SUCCESS in 3m 04s\n- openstack-tox-pep8 http://logs.openstack.org/33/634333/5/check/openstack-tox-pep8/aa888cf/ : SUCCESS in 2m 31s","accounts_in_message":[],"_revision_number":5},{"id":"403d5d5ca14ba322f17aaa69b70ec28d561a5d25","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2019-02-07 12:57:44.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nHi Eric,\n\nThanks for the updates, I like it much better this way. I added one more nit and a questions in line.\n\nThanks,\nIldikó","accounts_in_message":[],"_revision_number":5},{"id":"fe607fd5e10c8d4496db4af4838ce4b27153cba5","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-07 15:17:33.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"5115d36aff4d9545478f5bb5ce44e04a83754ac2","author":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"date":"2019-02-07 15:17:51.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"6ff148168f7106f381afcada676d7c753be4795a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-07 15:49:19.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/634333/6/check/openstack-tox-docs/04fbcc2/html/ : SUCCESS in 3m 56s\n- openstack-tox-pep8 http://logs.openstack.org/33/634333/6/check/openstack-tox-pep8/d52b0b4/ : SUCCESS in 3m 31s","accounts_in_message":[],"_revision_number":6},{"id":"023d746290e90552652e05f74b5a224d1b8a9e28","author":{"_account_id":9562,"name":"Ildiko Vancsa","email":"ildiko.vancsa@gmail.com","username":"ildikov"},"date":"2019-02-07 20:49:33.000000000","message":"Patch Set 6: Code-Review+2\n\nHi Eric,\n\nThanks for the fixes, I think it looks good now.\n\nBest Regards,\nIldikó","accounts_in_message":[],"_revision_number":6},{"id":"10ab9212b09d3226674a6371a99ec7d2c3af3182","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-02-11 21:04:44.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\nLooks to me like Ildiko covered everything already :) Thanks Eric!","accounts_in_message":[],"_revision_number":6},{"id":"66bbc33fa86f80676e13751d1d511d1d9e34a4d8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 21:04:56.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"2b1959e83b2b24e6a9af9f1648aac0c22ce392f0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 21:11:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"dafea95772c0e0c14c7a97c8645d2802447a4030","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-02-11 21:11:58.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/33/634333/6/gate/openstack-tox-docs/b200663/html/ : SUCCESS in 5m 27s\n- openstack-tox-pep8 http://logs.openstack.org/33/634333/6/gate/openstack-tox-pep8/77c4e66/ : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"7a2205a0c9b8ba626357de93d2f4a1f01e0ca998","revisions":{"390b8f417cd26bcfcb5b2c7a4d9e71c71065225b":{"kind":"REWORK","_number":1,"created":"2019-01-31 22:28:18.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/33/634333/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/contributor-guide","ref":"refs/changes/33/634333/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/1"}}},"commit":{"parents":[{"commit":"d4ecacdf9df700ce49d1966cf62e151bfcacdad9","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/d4ecacdf9df700ce49d1966cf62e151bfcacdad9"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-01-31 22:20:45.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-01-31 22:28:17.000000000","tz":-360},"subject":"Tutorial: changes in a series","message":"Tutorial: changes in a series\n\nThere was a mailing list post [1] walking through developing multiple\nchanges in a series. It was suggested [2] that this content would be\nsuitable for the contributor guide. This patch introduces a new document\nfor that, and also attempts to incorporate some of the other suggestions\n[3] [4] made in that thread.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000629.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000638.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000637.html\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000645.html\n\nChange-Id: I2e00505621ba4fec99e98bc1bfb1e37d2468d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/390b8f417cd26bcfcb5b2c7a4d9e71c71065225b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/390b8f417cd26bcfcb5b2c7a4d9e71c71065225b"}]},"branch":"refs/heads/master"},"65db25fef099efbdc068be636f56c208da56028e":{"kind":"REWORK","_number":2,"created":"2019-02-01 15:46:25.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/33/634333/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/contributor-guide","ref":"refs/changes/33/634333/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/2"}}},"commit":{"parents":[{"commit":"bd5f9d371fb9f7e96b4c3a44ad77b45199de0b94","subject":"Fix translation errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/bd5f9d371fb9f7e96b4c3a44ad77b45199de0b94"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-01-31 22:20:45.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-02-01 15:45:24.000000000","tz":-360},"subject":"Tutorial: changes in a series","message":"Tutorial: changes in a series\n\nThere was a mailing list post [1] walking through developing multiple\nchanges in a series. It was suggested [2] that this content would be\nsuitable for the contributor guide. This patch introduces a new document\nfor that, and also attempts to incorporate some of the other suggestions\n[3] [4] made in that thread.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000629.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000638.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000637.html\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000645.html\n\nChange-Id: I2e00505621ba4fec99e98bc1bfb1e37d2468d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/65db25fef099efbdc068be636f56c208da56028e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/65db25fef099efbdc068be636f56c208da56028e"}]},"branch":"refs/heads/master"},"19941379f5412a8d408027d6505da0eeeba6e76b":{"kind":"NO_CHANGE","_number":3,"created":"2019-02-01 16:02:57.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/33/634333/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/contributor-guide","ref":"refs/changes/33/634333/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/3"}}},"commit":{"parents":[{"commit":"176e0d1abef27e1a02e26075086e8ea32dbf6846","subject":"Fix translation errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/176e0d1abef27e1a02e26075086e8ea32dbf6846"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-01-31 22:20:45.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-02-01 16:02:11.000000000","tz":-360},"subject":"Tutorial: changes in a series","message":"Tutorial: changes in a series\n\nThere was a mailing list post [1] walking through developing multiple\nchanges in a series. It was suggested [2] that this content would be\nsuitable for the contributor guide. This patch introduces a new document\nfor that, and also attempts to incorporate some of the other suggestions\n[3] [4] made in that thread.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000629.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000638.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000637.html\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000645.html\n\nChange-Id: I2e00505621ba4fec99e98bc1bfb1e37d2468d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/19941379f5412a8d408027d6505da0eeeba6e76b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/19941379f5412a8d408027d6505da0eeeba6e76b"}]},"branch":"refs/heads/master"},"b51f95634437bcc999836bcbe2d15e61f7d99997":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2019-02-01 19:06:07.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/33/634333/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/contributor-guide","ref":"refs/changes/33/634333/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/4"}}},"commit":{"parents":[{"commit":"24f6e7cab26b6076a4e28e50fc2c8e5155032f6b","subject":"Don\u0027t error on translation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/24f6e7cab26b6076a4e28e50fc2c8e5155032f6b"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-01-31 22:20:45.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-02-01 19:05:50.000000000","tz":-360},"subject":"Tutorial: changes in a series","message":"Tutorial: changes in a series\n\nThere was a mailing list post [1] walking through developing multiple\nchanges in a series. It was suggested [2] that this content would be\nsuitable for the contributor guide. This patch introduces a new document\nfor that, and also attempts to incorporate some of the other suggestions\n[3] [4] made in that thread.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000629.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000638.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000637.html\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000645.html\n\nChange-Id: I2e00505621ba4fec99e98bc1bfb1e37d2468d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/b51f95634437bcc999836bcbe2d15e61f7d99997"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/b51f95634437bcc999836bcbe2d15e61f7d99997"}]},"branch":"refs/heads/master"},"ba8f81253ae250739712f88a07eef48ff97bc6df":{"kind":"REWORK","_number":5,"created":"2019-02-06 22:32:40.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/33/634333/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/contributor-guide","ref":"refs/changes/33/634333/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/5"}}},"commit":{"parents":[{"commit":"24f6e7cab26b6076a4e28e50fc2c8e5155032f6b","subject":"Don\u0027t error on translation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/24f6e7cab26b6076a4e28e50fc2c8e5155032f6b"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-01-31 22:20:45.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-02-06 22:32:29.000000000","tz":-360},"subject":"Tutorial: changes in a series","message":"Tutorial: changes in a series\n\nThere was a mailing list post [1] walking through developing multiple\nchanges in a series. It was suggested [2] that this content would be\nsuitable for the contributor guide. This patch introduces a new document\nfor that, and also attempts to incorporate some of the other suggestions\n[3] [4] made in that thread.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000629.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000638.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000637.html\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000645.html\n\nChange-Id: I2e00505621ba4fec99e98bc1bfb1e37d2468d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/ba8f81253ae250739712f88a07eef48ff97bc6df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/ba8f81253ae250739712f88a07eef48ff97bc6df"}]},"branch":"refs/heads/master"},"7a2205a0c9b8ba626357de93d2f4a1f01e0ca998":{"kind":"REWORK","_number":6,"created":"2019-02-07 15:17:51.000000000","uploader":{"_account_id":14070,"name":"Eric Fried","email":"openstack@fried.cc","username":"efried"},"ref":"refs/changes/33/634333/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/contributor-guide","ref":"refs/changes/33/634333/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/contributor-guide refs/changes/33/634333/6"}}},"commit":{"parents":[{"commit":"24f6e7cab26b6076a4e28e50fc2c8e5155032f6b","subject":"Don\u0027t error on translation warnings","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/24f6e7cab26b6076a4e28e50fc2c8e5155032f6b"}]}],"author":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-01-31 22:20:45.000000000","tz":-360},"committer":{"name":"Eric Fried","email":"openstack@fried.cc","date":"2019-02-07 15:17:49.000000000","tz":-360},"subject":"Tutorial: changes in a series","message":"Tutorial: changes in a series\n\nThere was a mailing list post [1] walking through developing multiple\nchanges in a series. It was suggested [2] that this content would be\nsuitable for the contributor guide. This patch introduces a new document\nfor that, and also attempts to incorporate some of the other suggestions\n[3] [4] made in that thread.\n\n[1] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000629.html\n[2] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000638.html\n[3] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000637.html\n[4] http://lists.openstack.org/pipermail/openstack-discuss/2018-December/000645.html\n\nChange-Id: I2e00505621ba4fec99e98bc1bfb1e37d2468d96d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/7a2205a0c9b8ba626357de93d2f4a1f01e0ca998"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/contributor-guide/commit/7a2205a0c9b8ba626357de93d2f4a1f01e0ca998"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
