)]}'
{"id":"openstack%2Fcookbook-openstack-bare-metal~697707","triplet_id":"openstack%2Fcookbook-openstack-bare-metal~master~Id39550d931218a60c5229802d9f660ce59ae0ad6","project":"openstack/cookbook-openstack-bare-metal","branch":"master","hashtags":[],"change_id":"Id39550d931218a60c5229802d9f660ce59ae0ad6","subject":"Updates for rocky","status":"MERGED","created":"2019-12-06 18:35:57.000000000","updated":"2019-12-08 10:24:20.000000000","submitted":"2019-12-08 10:24:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"697707-1575800660649-aff7a9a8","meta_rev_id":"9fc4ffad8baa65e65c3839e24bcf349f8ac42df9","_number":697707,"virtual_id_number":697707,"owner":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-08 10:24:20.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2019-12-08 09:14:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2019-12-08 09:14:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-08 09:14:36.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2019-12-08 10:24:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"77ff2e0b8876ad396c387c1814900ae545fa8585","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2019-12-06 18:35:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9904a4cd33d1efe6172f753e8b6a58296c317511","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 18:41:59.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/4eb09ce053ac489ab06a1dc8bdff1952 : FAILURE in 2m 52s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/4dd3fa39209e4e6db12705e31700b5de : POST_FAILURE in 2m 35s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/85611a08dc3c4a5bb88414ee8de9bec6 : POST_FAILURE in 3m 05s","accounts_in_message":[],"_revision_number":1},{"id":"0b55b3fb6f776c20e971b2f72c89f9c7ff93c49f","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2019-12-06 19:07:50.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"402f965202ce0481720ce7f4904d1f63111c143c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 19:57:52.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/32715d2cb2bc453699eadc90c90cc282 : FAILURE in 2m 48s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/5151e1cf374a4106888ed48654807aeb : SUCCESS in 42m 29s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/46b1339c06234d89be962acd30b7587d : SUCCESS in 46m 38s","accounts_in_message":[],"_revision_number":2},{"id":"6d0377f38b1b8c1c3697254227f21387d9ac8c5a","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2019-12-06 20:09:29.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dcd72ea367abec939a6377429702a06230115725","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-06 21:00:30.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/74d4f6d660b242e7a249e79ba27b74f6 : SUCCESS in 4m 22s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/85e9ffaa8b044cae9e585be2b80d1129 : SUCCESS in 40m 12s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/c3c2316094324a89afc2bbccffe20e56 : SUCCESS in 50m 09s","accounts_in_message":[],"_revision_number":3},{"id":"5c6092ad49c775f9b41e965a8f919644f00c2fd4","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-12-08 09:14:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3ccf810b5f7ded8043e50e203bbd5b8ae4001264","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 09:14:48.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"c0fe6d9c839487dd13f7fc1c0b13179ca1d0452c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 10:24:20.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"9fc4ffad8baa65e65c3839e24bcf349f8ac42df9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-08 10:24:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/dd2eee4f97084f9ca5b58bedb3dfb170 : SUCCESS in 3m 38s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/efc248733964472b9efedfc91b04e2ff : SUCCESS in 42m 31s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/7b94b06197a94680a9a0995f7c65e9c6 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"92b6694a0a97be485ca46d6c233e900525a498e3","revisions":{"1c8c9a669de5f3b3570a68d59c46c8bb990bc049":{"kind":"REWORK","_number":1,"created":"2019-12-06 18:35:57.000000000","uploader":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"ref":"refs/changes/07/697707/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-bare-metal","ref":"refs/changes/07/697707/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/1"}}},"commit":{"parents":[{"commit":"7ee0456dac423d0553da551c186b564bb5fd0a87","subject":"Use python3 packages on Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/7ee0456dac423d0553da551c186b564bb5fd0a87"}]}],"author":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2019-12-06 18:33:21.000000000","tz":-480},"committer":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2019-12-06 18:34:30.000000000","tz":-480},"subject":"Updates for rocky","message":"Updates for rocky\n\n- Replace git.openstack.org with opendev.org\n- Update some documentation\n- Move README.md to README.rst for better rendering\n\nChange-Id: Id39550d931218a60c5229802d9f660ce59ae0ad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/1c8c9a669de5f3b3570a68d59c46c8bb990bc049"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/1c8c9a669de5f3b3570a68d59c46c8bb990bc049"}]},"branch":"refs/heads/master"},"9d45f71d39b381b0903576b100f85804e37264da":{"kind":"REWORK","_number":2,"created":"2019-12-06 19:07:50.000000000","uploader":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"ref":"refs/changes/07/697707/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-bare-metal","ref":"refs/changes/07/697707/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/2"}}},"commit":{"parents":[{"commit":"7ee0456dac423d0553da551c186b564bb5fd0a87","subject":"Use python3 packages on Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/7ee0456dac423d0553da551c186b564bb5fd0a87"}]}],"author":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2019-12-06 18:33:21.000000000","tz":-480},"committer":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2019-12-06 19:07:46.000000000","tz":-480},"subject":"Updates for rocky","message":"Updates for rocky\n\n- Replace git.openstack.org with opendev.org\n- Update some documentation\n- Move README.md to README.rst for better rendering\n\nChange-Id: Id39550d931218a60c5229802d9f660ce59ae0ad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/9d45f71d39b381b0903576b100f85804e37264da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/9d45f71d39b381b0903576b100f85804e37264da"}]},"branch":"refs/heads/master"},"92b6694a0a97be485ca46d6c233e900525a498e3":{"kind":"REWORK","_number":3,"created":"2019-12-06 20:09:29.000000000","uploader":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"ref":"refs/changes/07/697707/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-bare-metal","ref":"refs/changes/07/697707/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-bare-metal refs/changes/07/697707/3"}}},"commit":{"parents":[{"commit":"7ee0456dac423d0553da551c186b564bb5fd0a87","subject":"Use python3 packages on Ubuntu","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/7ee0456dac423d0553da551c186b564bb5fd0a87"}]}],"author":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2019-12-06 18:33:21.000000000","tz":-480},"committer":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2019-12-06 20:09:13.000000000","tz":-480},"subject":"Updates for rocky","message":"Updates for rocky\n\n- Replace git.openstack.org with opendev.org\n- Update some documentation\n- Move README.md to README.rst for better rendering\n\nChange-Id: Id39550d931218a60c5229802d9f660ce59ae0ad6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/92b6694a0a97be485ca46d6c233e900525a498e3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-bare-metal/commit/92b6694a0a97be485ca46d6c233e900525a498e3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
