)]}'
{"id":"openstack%2Fcookbook-openstack-block-storage~574977","triplet_id":"openstack%2Fcookbook-openstack-block-storage~master~I188edfcfbd6f8261a648c03b6e46532b960d3ac6","project":"openstack/cookbook-openstack-block-storage","branch":"master","topic":"bp/simplify-identity-endpoint","hashtags":[],"change_id":"I188edfcfbd6f8261a648c03b6e46532b960d3ac6","subject":"Simplify identity endpoint","status":"MERGED","created":"2018-06-13 04:48:11.000000000","updated":"2018-06-28 15:38:32.000000000","submitted":"2018-06-28 15:38:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"574977-1530200312647-60039b84","meta_rev_id":"df5fc01cd098b1c0cab02365bfc593e5ab129b4f","_number":574977,"virtual_id_number":574977,"owner":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":2,"date":"2018-06-28 15:38:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"all":[{"value":0,"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":2,"date":"2018-06-26 13:33:53.000000000","_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"all":[{"value":1,"date":"2018-06-28 14:34:52.000000000","_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-06-26 13:33:53.000000000","updated_by":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"reviewer":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"state":"REVIEWER"},{"updated":"2018-06-28 14:34:52.000000000","updated_by":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"reviewer":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"state":"REVIEWER"},{"updated":"2018-06-28 15:38:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"105817596956532b260c591b62a5e1c18bc20daf","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2018-06-13 04:48:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f5ce745e73f22f67c4e7fcc6a47ef5ed18a12757","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2018-06-13 04:50:40.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"0780f03e969109bd1f34e451a621f7784382207d","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-13 05:30:21.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/77/574977/2/check/openstack-chef-repo-rake/9efdd13/ : FAILURE in 8m 55s\n- openstack-chef-repo-integration http://logs.openstack.org/77/574977/2/check/openstack-chef-repo-integration/49592c6/ : SUCCESS in 38m 11s","accounts_in_message":[],"_revision_number":2},{"id":"74bc46a2cfea112952428fc5078d67ee41998b16","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2018-06-13 06:00:00.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"ce9d25ae7b3650b766bfe83d17a67351d2a12800","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-13 06:17:20.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/77/574977/3/check/openstack-chef-repo-rake/94bc350/ : FAILURE in 10m 00s\n- openstack-chef-repo-integration http://logs.openstack.org/77/574977/3/check/openstack-chef-repo-integration/9fe6304/ : FAILURE in 6m 12s","accounts_in_message":[],"_revision_number":3},{"id":"93f176c86984b01f4e85fde49340ed562acc32fc","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2018-06-15 02:14:59.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"77c121fb832b4c6719c74683c0f271764bda48a6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-15 03:48:17.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/77/574977/4/check/openstack-chef-repo-rake/9dca433/ : FAILURE in 20m 02s\n- openstack-chef-repo-integration http://logs.openstack.org/77/574977/4/check/openstack-chef-repo-integration/0e00920/ : SUCCESS in 39m 52s","accounts_in_message":[],"_revision_number":4},{"id":"a9d87d587c6e0f79acfa6be3cc78e713e05b4151","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2018-06-15 04:28:50.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"768f8e2869b5e9f18ecdbd81960ad2d4d1064f03","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-15 05:51:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-repo-rake http://logs.openstack.org/77/574977/5/check/openstack-chef-repo-rake/7646162/ : SUCCESS in 10m 28s\n- openstack-chef-repo-integration http://logs.openstack.org/77/574977/5/check/openstack-chef-repo-integration/e724687/ : SUCCESS in 36m 35s","accounts_in_message":[],"_revision_number":5},{"id":"e32a78e26392a33c9fc3217a4efd31047431491a","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2018-06-26 13:33:53.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"4a96e07bea94a0c8370d4143f8a889b53653f9eb","author":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"date":"2018-06-28 14:34:52.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"10d36677341be6119d9ac8bd8766fc2739b87ddf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 14:44:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"0222ac910b4020530ff8f4497f032719c956e00a","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 15:38:32.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"5313075f4b54102245edd2b3a075d47397910461","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-06-28 15:38:32.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-repo-rake http://logs.openstack.org/77/574977/5/gate/openstack-chef-repo-rake/e20abda/ : SUCCESS in 10m 44s\n- openstack-chef-repo-integration http://logs.openstack.org/77/574977/5/gate/openstack-chef-repo-integration/94449b9/ : SUCCESS in 36m 41s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"93882de60329d91fdb9ba6ad5515a94312f570a8","revisions":{"5f70ed8c7679da08cd659645f874b036dd026940":{"kind":"REWORK","_number":1,"created":"2018-06-13 04:48:11.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/77/574977/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-block-storage","ref":"refs/changes/77/574977/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/1"}}},"commit":{"parents":[{"commit":"c26101bfbad23f19dddb5b31a1937490495ff569","subject":"Add delivery config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/c26101bfbad23f19dddb5b31a1937490495ff569"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 04:04:02.000000000","tz":-420},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 04:48:10.000000000","tz":-420},"subject":"Simplify identity endpoint","message":"Simplify identity endpoint\n\nPer the Keystone Install Guide[1] the admin endpoint is superseded in\nfavor of a single public endpoint. As a result, the admin endpoint is no\nlonger deployed by default.\n\n[1] https://docs.openstack.org/keystone/queens/install/keystone-install-ubuntu.html#install-and-configure-components\n\nChange-Id: I188edfcfbd6f8261a648c03b6e46532b960d3ac6\nImplements: bp/simplify-identity-endpoint\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/5f70ed8c7679da08cd659645f874b036dd026940"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/5f70ed8c7679da08cd659645f874b036dd026940"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c26101bfbad23f19dddb5b31a1937490495ff569","is_merged_in_target_branch":true,"change_id":"Ia6cdd3dfb6ca7db2622caaf89821a1ed4c3e76f1","change_number":560753,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7b57fd2d35c70a99aecf25880bd1e17b8e38ece4":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2018-06-13 04:50:40.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/77/574977/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-block-storage","ref":"refs/changes/77/574977/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/2"}}},"commit":{"parents":[{"commit":"c26101bfbad23f19dddb5b31a1937490495ff569","subject":"Add delivery config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/c26101bfbad23f19dddb5b31a1937490495ff569"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 04:04:02.000000000","tz":-420},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 04:50:26.000000000","tz":-420},"subject":"Simplify identity endpoint","message":"Simplify identity endpoint\n\nPer the Keystone Install Guide[1] the admin endpoint is superseded in\nfavor of a single public endpoint. As a result, the admin endpoint is no\nlonger deployed by default.\n\n[1] https://docs.openstack.org/keystone/queens/install/keystone-install-ubuntu.html#install-and-configure-components\n\nChange-Id: I188edfcfbd6f8261a648c03b6e46532b960d3ac6\nImplements: blueprint simplify-identity-endpoint\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/7b57fd2d35c70a99aecf25880bd1e17b8e38ece4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/7b57fd2d35c70a99aecf25880bd1e17b8e38ece4"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c26101bfbad23f19dddb5b31a1937490495ff569","is_merged_in_target_branch":true,"change_id":"Ia6cdd3dfb6ca7db2622caaf89821a1ed4c3e76f1","change_number":560753,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"91d1672d3eddd2be49d0ef2b194896ece185ff39":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2018-06-13 06:00:00.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/77/574977/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-block-storage","ref":"refs/changes/77/574977/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/3"}}},"commit":{"parents":[{"commit":"c26101bfbad23f19dddb5b31a1937490495ff569","subject":"Add delivery config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/c26101bfbad23f19dddb5b31a1937490495ff569"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 04:04:02.000000000","tz":-420},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 05:59:57.000000000","tz":0},"subject":"Simplify identity endpoint","message":"Simplify identity endpoint\n\nPer the Keystone Install Guide[1] the admin endpoint is superseded in\nfavor of a single public endpoint. As a result, the admin endpoint is no\nlonger deployed by default.\n\n[1] https://docs.openstack.org/keystone/queens/install/keystone-install-ubuntu.html#install-and-configure-components\n\nDepends-On: Ied0fb46ae8c10273fde31691b910dc2748845faf\nChange-Id: I188edfcfbd6f8261a648c03b6e46532b960d3ac6\nImplements: blueprint simplify-identity-endpoint\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/91d1672d3eddd2be49d0ef2b194896ece185ff39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/91d1672d3eddd2be49d0ef2b194896ece185ff39"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c26101bfbad23f19dddb5b31a1937490495ff569","is_merged_in_target_branch":true,"change_id":"Ia6cdd3dfb6ca7db2622caaf89821a1ed4c3e76f1","change_number":560753,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"78929c500dfa2c2735b1d9d4b81ff80dd2368f4c":{"kind":"REWORK","_number":4,"created":"2018-06-15 02:14:59.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/77/574977/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-block-storage","ref":"refs/changes/77/574977/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/4"}}},"commit":{"parents":[{"commit":"c26101bfbad23f19dddb5b31a1937490495ff569","subject":"Add delivery config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/c26101bfbad23f19dddb5b31a1937490495ff569"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 04:04:02.000000000","tz":-420},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-15 02:14:55.000000000","tz":-420},"subject":"Simplify identity endpoint","message":"Simplify identity endpoint\n\nPer the Keystone Install Guide[1] the admin endpoint is superseded in\nfavor of a single public endpoint. As a result, the admin endpoint is no\nlonger deployed by default.\n\n[1] https://docs.openstack.org/keystone/queens/install/keystone-install-ubuntu.html#install-and-configure-components\n\nChange-Id: I188edfcfbd6f8261a648c03b6e46532b960d3ac6\nImplements: blueprint simplify-identity-endpoint\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/78929c500dfa2c2735b1d9d4b81ff80dd2368f4c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/78929c500dfa2c2735b1d9d4b81ff80dd2368f4c"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c26101bfbad23f19dddb5b31a1937490495ff569","is_merged_in_target_branch":true,"change_id":"Ia6cdd3dfb6ca7db2622caaf89821a1ed4c3e76f1","change_number":560753,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"93882de60329d91fdb9ba6ad5515a94312f570a8":{"kind":"REWORK","_number":5,"created":"2018-06-15 04:28:50.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/77/574977/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-block-storage","ref":"refs/changes/77/574977/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-block-storage refs/changes/77/574977/5"}}},"commit":{"parents":[{"commit":"c26101bfbad23f19dddb5b31a1937490495ff569","subject":"Add delivery config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/c26101bfbad23f19dddb5b31a1937490495ff569"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-13 04:04:02.000000000","tz":-420},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2018-06-15 04:28:42.000000000","tz":-420},"subject":"Simplify identity endpoint","message":"Simplify identity endpoint\n\nPer the Keystone Install Guide[1] the admin endpoint is superseded in\nfavor of a single public endpoint. As a result, the admin endpoint is no\nlonger deployed by default.\n\n[1] https://docs.openstack.org/keystone/queens/install/keystone-install-ubuntu.html#install-and-configure-components\n\nChange-Id: I188edfcfbd6f8261a648c03b6e46532b960d3ac6\nImplements: blueprint simplify-identity-endpoint\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/93882de60329d91fdb9ba6ad5515a94312f570a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-block-storage/commit/93882de60329d91fdb9ba6ad5515a94312f570a8"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c26101bfbad23f19dddb5b31a1937490495ff569","is_merged_in_target_branch":true,"change_id":"Ia6cdd3dfb6ca7db2622caaf89821a1ed4c3e76f1","change_number":560753,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
