)]}'
{"id":"openstack%2Fcookbook-openstack-dashboard~494503","triplet_id":"openstack%2Fcookbook-openstack-dashboard~master~I261b555fa826e83c17ba61e734cbeb343035b667","project":"openstack/cookbook-openstack-dashboard","branch":"master","hashtags":[],"change_id":"I261b555fa826e83c17ba61e734cbeb343035b667","subject":"starting pike development patch","status":"MERGED","created":"2017-08-17 12:19:12.000000000","updated":"2017-08-18 21:43:57.000000000","submitted":"2017-08-18 21:33:00.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"501fff62a76bd321711c4578f54fd1435b39bd4d","_number":494503,"virtual_id_number":494503,"owner":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-08-18 21:33:00.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":0,"date":"2017-08-18 21:43:57.000000000","post_submit":true,"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-08-18 21:18:07.000000000","_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":0,"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-08-18 21:18:07.000000000","_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":0,"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-18 21:18:07.000000000","updated_by":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"reviewer":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"state":"REVIEWER"},{"updated":"2017-08-18 21:33:00.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"},{"updated":"2017-08-18 21:43:57.000000000","updated_by":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"reviewer":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"state":"REVIEWER"}],"messages":[{"id":"c5e31c372ea342902d1f26f04508d749f2605ba6","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2017-08-17 12:19:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e22c36f982890188fa0026280978e63d15d0cd38","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-17 12:29:05.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-cookbook-openstack-dashboard-chef-rake http://logs.openstack.org/03/494503/1/check/gate-cookbook-openstack-dashboard-chef-rake/f1bfcad/ : FAILURE in 7m 41s\n- gate-cookbook-openstack-dashboard-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/03/494503/1/check/gate-cookbook-openstack-dashboard-chef-rake-integration-ubuntu-xenial-nv/acf576a/ : FAILURE in 4m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e9886cac3392db7fd8ff2425bd9d10f28dfd4a7","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-08-18 19:48:38.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"4c1dd164a54b9790a3fe996e19410cb7653b5e1c","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 21:16:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cookbook-openstack-dashboard-chef-rake http://logs.openstack.org/03/494503/1/check/gate-cookbook-openstack-dashboard-chef-rake/a65160b/ : SUCCESS in 17m 00s\n- gate-cookbook-openstack-dashboard-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/03/494503/1/check/gate-cookbook-openstack-dashboard-chef-rake-integration-ubuntu-xenial-nv/966704e/ : FAILURE in 34m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bf87c7b0cd2875981bfa09a56bd42137f1b164ff","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-08-18 21:18:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"70559641b4bbe500667df60366bea1ed115b2b8a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 21:18:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":1},{"id":"cb3a0b8e87348b819bb2100803cfb14639e9403f","date":"2017-08-18 21:33:00.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":1},{"id":"80be041fa8ae3e047e29dc33c5a9f8804acd4102","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 21:33:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-cookbook-openstack-dashboard-chef-rake http://logs.openstack.org/03/494503/1/gate/gate-cookbook-openstack-dashboard-chef-rake/58a106f/ : SUCCESS in 14m 29s","accounts_in_message":[],"_revision_number":1},{"id":"501fff62a76bd321711c4578f54fd1435b39bd4d","author":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"date":"2017-08-18 21:43:57.000000000","message":"Patch Set 1:\n\n* cloudbau-Chef-OpenStack-CI http://ci.cloudbau.de/ci-logs/refs-changes-03-494503-1 : SUCCESS","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e44af9d178cebf882c30c92a8f15f5ba4ce5474","revisions":{"6e44af9d178cebf882c30c92a8f15f5ba4ce5474":{"kind":"REWORK","_number":1,"created":"2017-08-17 12:19:12.000000000","uploader":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"ref":"refs/changes/03/494503/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-dashboard","ref":"refs/changes/03/494503/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/03/494503/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/03/494503/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/03/494503/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/03/494503/1"}}},"commit":{"parents":[{"commit":"5c5d37fe5594170e59bd000a3fca7bc232b4c937","subject":"Style and lint fixes for the Ocata release","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/5c5d37fe5594170e59bd000a3fca7bc232b4c937"}]}],"author":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-17 12:19:07.000000000","tz":120},"committer":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-17 12:19:07.000000000","tz":120},"subject":"starting pike development patch","message":"starting pike development patch\n\nChange-Id: I261b555fa826e83c17ba61e734cbeb343035b667\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/6e44af9d178cebf882c30c92a8f15f5ba4ce5474"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/6e44af9d178cebf882c30c92a8f15f5ba4ce5474"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5c5d37fe5594170e59bd000a3fca7bc232b4c937","is_merged_in_target_branch":true,"change_id":"I63e6680cec8b66e2ece2d2627c0b413f5d401317","change_number":489863,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
