)]}'
{"id":"openstack%2Fcookbook-openstack-dashboard~750029","triplet_id":"openstack%2Fcookbook-openstack-dashboard~stable%2Fstein~Ief6efb3d2afe2af560fa76eea5fc0b90cadaee7c","project":"openstack/cookbook-openstack-dashboard","branch":"stable/stein","topic":"chef16-stein","hashtags":[],"change_id":"Ief6efb3d2afe2af560fa76eea5fc0b90cadaee7c","subject":"Chef 16 updates","status":"MERGED","created":"2020-09-04 20:04:15.000000000","updated":"2020-09-09 17:01:40.000000000","submitted":"2020-09-09 17:01:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"750029-1599670900548-7f25b096","meta_rev_id":"70748f46b097eef2d7999e8008f974729e7799cf","_number":750029,"virtual_id_number":750029,"owner":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-09 17:01:40.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2020-09-08 05:13:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":1,"date":"2020-09-09 15:51:18.000000000","_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-09-08 05:13:42.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2020-09-09 17:01:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"943390fc1e2d32a6f60876df0b126a0dea3517dd","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2020-09-04 20:04:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37387f19a52f462dfb2e5b2dfdf94733532f032a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-04 20:51:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/4083ccef60f54ed0907f958551c3c58a : SUCCESS in 4m 31s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/24639f154c254f38881bfc68937f5ed3 : SUCCESS in 15m 35s (non-voting)\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/6a89542564c742579fe73e4a29021e89 : SUCCESS in 13m 37s (non-voting)\n- openstack-chef-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/4c90fe122d50460f8d8517d294bffadd : SUCCESS in 45m 11s (non-voting)\n- openstack-chef-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/21f1b017a030448eb5627ae8f1df43c3 : SUCCESS in 40m 13s (non-voting)\n- openstack-chef15-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/4ee8e9f074374e69996dabc23e9fd0df : SUCCESS in 43m 43s (non-voting)\n- openstack-chef15-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/14703dd303c444d7b3c89dc9fc7c0b63 : SUCCESS in 44m 12s (non-voting)\n- openstack-cinc15-integration-centos https://zuul.opendev.org/t/openstack/build/e48cdf5b501d4ae0b5996b6f74d15639 : SUCCESS in 13m 13s (non-voting)\n- openstack-cinc15-integration-ubuntu https://zuul.opendev.org/t/openstack/build/6bf9cb9953984a5491371fa2bf4b5946 : SUCCESS in 15m 01s (non-voting)\n- openstack-cinc-integration-centos https://zuul.opendev.org/t/openstack/build/d2e68d6f4af845cda520f80336e62f85 : SUCCESS in 13m 45s (non-voting)\n- openstack-cinc-integration-ubuntu https://zuul.opendev.org/t/openstack/build/f52c906f931b424fb27d2f7e790ec3dc : SUCCESS in 11m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7f60661de9e7acb7ce989cb1de885a84c64c47c","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2020-09-08 05:13:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"1334579bbd3a272bc3fa19aa06f29e8c2b0a026c","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2020-09-09 15:51:18.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8fb64096c14ac4528b18315bd7dd6649779e5a48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 15:55:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"96405a9e9156bdb97f306902f55a226a3bd3d550","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 17:01:40.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"d6d297fd7eca93ee0a338c3a20e0dc843904e423","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-09 17:01:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/0f01df4e675b4bc29de7256147deb341 : SUCCESS in 5m 54s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/be870a43ac244a5ca35828f72f2258a0 : SUCCESS in 12m 57s (non-voting)\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/283d9278fabe4170a111fd75193fb5da : SUCCESS in 14m 12s (non-voting)\n- openstack-chef-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/5942e89dfcc6405a9b2c83150126a533 : SUCCESS in 32m 53s (non-voting)\n- openstack-chef-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/68d7633ce3a54ef28370c0e040e085ac : SUCCESS in 40m 24s (non-voting)\n- openstack-chef15-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/4d6f3ffa4d374f18943927d12870bcfe : SUCCESS in 34m 20s (non-voting)\n- openstack-chef15-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/8c2f325a99c14608b97a04bdea8690da : SUCCESS in 41m 14s (non-voting)\n- openstack-cinc15-integration-centos https://zuul.opendev.org/t/openstack/build/176b065471074b20a3d86f362eae7cd4 : SUCCESS in 19m 04s (non-voting)\n- openstack-cinc15-integration-ubuntu https://zuul.opendev.org/t/openstack/build/e5f25bbbca334aa7b6a57a7aa12e9950 : SUCCESS in 18m 26s (non-voting)\n- openstack-cinc-integration-centos https://zuul.opendev.org/t/openstack/build/b79030c404bc43e499b1f40902865227 : SUCCESS in 15m 27s (non-voting)\n- openstack-cinc-integration-ubuntu https://zuul.opendev.org/t/openstack/build/58b8869985a648cc8dfbe827d3663181 : SUCCESS in 16m 24s (non-voting)","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"51f22d76233a6cf4758364f34812f4b5c98b5924","revisions":{"51f22d76233a6cf4758364f34812f4b5c98b5924":{"kind":"REWORK","_number":1,"created":"2020-09-04 20:04:15.000000000","uploader":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"ref":"refs/changes/29/750029/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-dashboard","ref":"refs/changes/29/750029/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/29/750029/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/29/750029/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/29/750029/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/29/750029/1"}}},"commit":{"parents":[{"commit":"847ab6a6fd501f55ef6245ca4561f707fb49385b","subject":"stable/stein release patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/847ab6a6fd501f55ef6245ca4561f707fb49385b"}]}],"author":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2020-08-24 15:46:42.000000000","tz":-420},"committer":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2020-09-04 20:03:55.000000000","tz":-420},"subject":"Chef 16 updates","message":"Chef 16 updates\n\nDepends-On: https://review.opendev.org/749849\nChange-Id: Ief6efb3d2afe2af560fa76eea5fc0b90cadaee7c\nSigned-off-by: Lance Albertson \u003clance@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/51f22d76233a6cf4758364f34812f4b5c98b5924"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/51f22d76233a6cf4758364f34812f4b5c98b5924"}]},"branch":"refs/heads/stable/stein"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
