)]}'
{"id":"openstack%2Fcookbook-openstack-dashboard~834250","triplet_id":"openstack%2Fcookbook-openstack-dashboard~stable%2Ftrain~I5a0053c729b99fc758d7d2a7b9831476e3ba373e","project":"openstack/cookbook-openstack-dashboard","branch":"stable/train","hashtags":[],"change_id":"I5a0053c729b99fc758d7d2a7b9831476e3ba373e","subject":"stable/train release patch","status":"MERGED","created":"2022-03-18 00:09:17.000000000","updated":"2022-03-18 17:29:09.000000000","submitted":"2022-03-18 17:29:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"834250","meta_rev_id":"f97c4b503850a7d006779a73ec653981e8263ddb","_number":834250,"virtual_id_number":834250,"owner":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-18 17:29:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"all":[{"value":2,"date":"2022-03-18 16:28:07.000000000","_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"all":[{"value":1,"date":"2022-03-18 16:28:07.000000000","_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-18 01:04:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0bfd3f5ab831acdca892adf596cdcb6cbab13883","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2022-03-18 00:09:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"181aa5ec35b121ff512ff92e63338ab18162fa0c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 01:04:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/144b19d3ccdf433ab0db72191812fcb9 : SUCCESS in 6m 17s (non-voting)\n- openstack-chef-integration-centos-7 https://zuul.opendev.org/t/openstack/build/c3e537281fb441dd8a6c43e4e3e7df74 : FAILURE in 6m 06s (non-voting)\n- openstack-chef-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/567b9801cbf94af99c48dfa833060061 : SUCCESS in 24m 06s (non-voting)\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/eaba7c115e6e4ea296c6c324fa963236 : SUCCESS in 14m 15s (non-voting)\n- openstack-chef-minimal-integration-centos-7 https://zuul.opendev.org/t/openstack/build/dfe132dccd234178ad6db92c4f71a5b4 : SUCCESS in 34m 46s (non-voting)\n- openstack-chef-minimal-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/658fc4cfc7f14fc8b4170df448f3d1a1 : SUCCESS in 48m 28s (non-voting)\n- openstack-chef-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/2299da467e514901869dd687e45f3f78 : SUCCESS in 32m 19s (non-voting)\n- openstack-chef16-minimal-integration-centos-7 https://zuul.opendev.org/t/openstack/build/280c9d401bef4053b93371db28338fe3 : SUCCESS in 43m 29s (non-voting)\n- openstack-chef16-minimal-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/a1d96cb0e4724351b6c9c7bbf9f9f086 : SUCCESS in 46m 09s (non-voting)\n- openstack-chef16-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/0418d9228301412cb06a4186aec88be7 : SUCCESS in 36m 35s (non-voting)\n- openstack-cinc16-integration-centos-7 https://zuul.opendev.org/t/openstack/build/2355b3bd9e0d4411bac03fef134de78d : SUCCESS in 17m 52s (non-voting)\n- openstack-cinc16-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/d8146bb973914583ada1b9855706878d : SUCCESS in 23m 20s (non-voting)\n- openstack-cinc16-integration-ubuntu https://zuul.opendev.org/t/openstack/build/22a1c196886d4681b0ae8845db415e30 : SUCCESS in 13m 20s (non-voting)\n- openstack-cinc-integration-centos-7 https://zuul.opendev.org/t/openstack/build/3aa57da013284801b4305701bd096f9f : SUCCESS in 20m 51s (non-voting)\n- openstack-cinc-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/32a2855cdb9340d0910b278ad23b0b82 : SUCCESS in 24m 43s (non-voting)\n- openstack-cinc-integration-ubuntu https://zuul.opendev.org/t/openstack/build/622f6b56d746408fa9365763d62c48f9 : SUCCESS in 13m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f09590e42fab5486cc3f90e3aded6d4913c49b4b","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2022-03-18 16:28:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"64bf6126ce1280bcafe6dff3a35960a979a06beb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 16:29:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22e09978f130c404f7a14bb24e75d1a6e9edcdb2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 17:29:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/0c38e23ffc1d4971a042bd93389b5492 : SUCCESS in 7m 23s (non-voting)\n- openstack-chef-integration-centos-7 https://zuul.opendev.org/t/openstack/build/c3a832a2956f435ab5258f0f9a166b41 : SUCCESS in 17m 26s (non-voting)\n- openstack-chef-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/939333799ac04b32a1d482d7e54321d1 : SUCCESS in 29m 35s (non-voting)\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/ad40d4e7c6ae43cd9fb94ad29949c8d9 : SUCCESS in 13m 18s (non-voting)\n- openstack-chef-minimal-integration-centos-7 https://zuul.opendev.org/t/openstack/build/b66e5827ba05441c83b08f9eabd279ff : SUCCESS in 49m 32s (non-voting)\n- openstack-chef-minimal-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/1b2af8770a7d459e8ce169b8748dc160 : SUCCESS in 48m 45s (non-voting)\n- openstack-chef-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/1e4028a2ac024628bdaa1ab4a53b03e1 : SUCCESS in 35m 37s (non-voting)\n- openstack-chef16-minimal-integration-centos-7 https://zuul.opendev.org/t/openstack/build/fdaae53d79d041cab9d4cbb1b37c5da1 : SUCCESS in 44m 53s (non-voting)\n- openstack-chef16-minimal-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/6c2daa1e54ad4d8898177443ea303b8e : SUCCESS in 52m 50s (non-voting)\n- openstack-chef16-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/a3dbf61658f04996ad02401905bd5573 : SUCCESS in 34m 46s (non-voting)\n- openstack-cinc16-integration-centos-7 https://zuul.opendev.org/t/openstack/build/df0f55dfb0aa44bba9005ddc714c8599 : SUCCESS in 20m 03s (non-voting)\n- openstack-cinc16-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/7028313cd6ab4848bce2747154fe0d9f : SUCCESS in 23m 53s (non-voting)\n- openstack-cinc16-integration-ubuntu https://zuul.opendev.org/t/openstack/build/664720780ae24f938dfc64a3d2c33214 : SUCCESS in 13m 43s (non-voting)\n- openstack-cinc-integration-centos-7 https://zuul.opendev.org/t/openstack/build/df4bd30533da484594c0aa7d34b863fe : SUCCESS in 17m 10s (non-voting)\n- openstack-cinc-integration-centos-stream-8 https://zuul.opendev.org/t/openstack/build/c27e45a3203144a29dc01844c23468e0 : SUCCESS in 24m 31s (non-voting)\n- openstack-cinc-integration-ubuntu https://zuul.opendev.org/t/openstack/build/d659d5818dd14ee396628659da6459e6 : SUCCESS in 15m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f97c4b503850a7d006779a73ec653981e8263ddb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-18 17:29:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"49857091afa929472181c975be6c68dede9d2e53","revisions":{"49857091afa929472181c975be6c68dede9d2e53":{"kind":"REWORK","_number":1,"created":"2022-03-18 00:09:17.000000000","uploader":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"ref":"refs/changes/50/834250/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-dashboard","ref":"refs/changes/50/834250/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/50/834250/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/50/834250/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/50/834250/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-dashboard refs/changes/50/834250/1"}}},"commit":{"parents":[{"commit":"de8fb53dc0363230ce86b0d9b2c6f724a999dcc2","subject":"CentOS 8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/de8fb53dc0363230ce86b0d9b2c6f724a999dcc2"}]}],"author":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2022-03-18 00:09:05.000000000","tz":-420},"committer":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2022-03-18 00:09:05.000000000","tz":-420},"subject":"stable/train release patch","message":"stable/train release patch\n\nDepends-On: https://review.opendev.org/c/openstack/cookbook-openstack-identity/+/834238\nSigned-off-by: Lance Albertson \u003clance@osuosl.org\u003e\nChange-Id: I5a0053c729b99fc758d7d2a7b9831476e3ba373e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/49857091afa929472181c975be6c68dede9d2e53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-dashboard/commit/49857091afa929472181c975be6c68dede9d2e53"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
