)]}'
{"id":"openstack%2Fcookbook-openstack-identity~909163","triplet_id":"openstack%2Fcookbook-openstack-identity~master~Ida0639315944c8c7852ec37fb10f133e8ab9c455","project":"openstack/cookbook-openstack-identity","branch":"master","topic":"retire-openstack-chef","attention_set":{},"removed_from_attention_set":{"11975":{"account":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"last_update":"2024-02-19 18:29:03.000000000","reason":"Change was submitted"},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-02-19 18:29:03.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ida0639315944c8c7852ec37fb10f133e8ab9c455","subject":"Retire openstack-chef: remove repo content","status":"MERGED","created":"2024-02-15 22:23:29.000000000","updated":"2024-02-19 18:29:03.000000000","submitted":"2024-02-19 18:29:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"909163-retire-openstack-chef","meta_rev_id":"afaa2f06232b3b2c31e46e7518255fce24342746","_number":909163,"virtual_id_number":909163,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-19 18:29:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":2,"date":"2024-02-19 18:28:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2024-02-19 16:58:25.000000000","_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":1,"date":"2024-02-18 18:21:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2024-02-19 18:28:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-16 22:45:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-17 11:52:05.000000000","updated_by":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"reviewer":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"state":"REVIEWER"},{"updated":"2024-02-18 18:21:15.000000000","updated_by":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"reviewer":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"state":"REVIEWER"},{"updated":"2024-02-19 16:58:25.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-02-19 18:28:11.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"}],"messages":[{"id":"01100cdeb6c2a7dd46ca8568a5aaf0073f73a1de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-15 22:23:29.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8822158f151316011c2dc72f5ceee921224a3e0f","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-16 22:44:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b6e394c368ef7e5034a7af9aad4765880a3c582f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 22:45:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/450e970fe9cc405d99c0bdbb40aa6ac6\n\n- noop https://zuul.opendev.org/t/openstack/build/fd08eab424704dada63ba1b00321632c : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"2ff21f6f4ff84803339d92e9fcf52e140c78bd36","author":{"_account_id":11975,"name":"Slawek Kaplonski","email":"skaplons@redhat.com","username":"slaweq"},"date":"2024-02-17 11:52:05.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"03490c60b9c5534e442963ce0fbccaa6e194ee3b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-02-18 04:58:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"2497a88bdf99f5d1ec8476efcdedafcecd79d706","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-18 04:58:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6d8bdcb9ed24a0d9871bbb7752fd6d2\n\n- noop https://zuul.opendev.org/t/openstack/build/82363a23e2ae424699c64be6c3e5b492 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"efcfc24b84b8548ccdfb3eee179bbc118053cb4c","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2024-02-18 18:21:15.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"12afc4878881fd4c3941c3054930e8ea394359cc","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-02-19 16:58:25.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"9d76f93516106b6382800698fdd87e35bacb50d1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-02-19 18:28:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"7bedabf27bb71148a916a28af40aaee7ce6ac08c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 18:28:53.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"6dc66897d19d5b9f69f0e6c88c9823c5fc4f0f37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 18:29:03.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aebb096d36884e2d98e8db9bf3b28829\n\n- noop https://zuul.opendev.org/t/openstack/build/093751ecc8224a31b3c038de09adae90 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"afaa2f06232b3b2c31e46e7518255fce24342746","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-19 18:29:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"44d13c8c6421d123a51e70562a5e43f4018b66de","revisions":{"d0c50225c39826e4deab54734e28301e88ed41a4":{"kind":"REWORK","_number":1,"created":"2024-02-15 22:23:29.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/63/909163/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-identity","ref":"refs/changes/63/909163/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/1"}}},"commit":{"parents":[{"commit":"f052ede42bab11320fefcafd8e3565269cc00dfb","subject":"CentOS 8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-identity/commit/f052ede42bab11320fefcafd8e3565269cc00dfb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-15 22:17:05.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-15 22:23:26.000000000","tz":-480},"subject":"Retire openstack-chef: remove repo content","message":"Retire openstack-chef: remove repo content\n\nOpenStack-chef project is retiring\n- https://review.opendev.org/c/openstack/governance/+/905279\n\nthis commit remove the content of this project repo\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/909134\nChange-Id: Ida0639315944c8c7852ec37fb10f133e8ab9c455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-identity/commit/d0c50225c39826e4deab54734e28301e88ed41a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-identity/commit/d0c50225c39826e4deab54734e28301e88ed41a4"}]},"branch":"refs/heads/master"},"44d13c8c6421d123a51e70562a5e43f4018b66de":{"kind":"REWORK","_number":2,"created":"2024-02-18 04:58:22.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/63/909163/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-identity","ref":"refs/changes/63/909163/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-identity refs/changes/63/909163/2"}}},"commit":{"parents":[{"commit":"f052ede42bab11320fefcafd8e3565269cc00dfb","subject":"CentOS 8 support","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-identity/commit/f052ede42bab11320fefcafd8e3565269cc00dfb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-15 22:17:05.000000000","tz":-480},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-02-18 04:50:52.000000000","tz":-480},"subject":"Retire openstack-chef: remove repo content","message":"Retire openstack-chef: remove repo content\n\nOpenStack-chef project is retiring\n- https://review.opendev.org/c/openstack/governance/+/905279\n\nthis commit remove the content of this project repo\n\nDepends-On: https://review.opendev.org/c/openstack/project-config/+/909134\nChange-Id: Ida0639315944c8c7852ec37fb10f133e8ab9c455\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-identity/commit/44d13c8c6421d123a51e70562a5e43f4018b66de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-identity/commit/44d13c8c6421d123a51e70562a5e43f4018b66de"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
