)]}'
{"id":"openstack%2Fcookbook-openstack-ops-messaging~494509","triplet_id":"openstack%2Fcookbook-openstack-ops-messaging~master~I76c914f68612c0e76dcee8e0557ff62f769533db","project":"openstack/cookbook-openstack-ops-messaging","branch":"master","topic":"494509","hashtags":[],"change_id":"I76c914f68612c0e76dcee8e0557ff62f769533db","subject":"starting pike development patch","status":"MERGED","created":"2017-08-17 12:19:52.000000000","updated":"2017-08-18 19:37:34.000000000","submitted":"2017-08-18 19:37:34.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"96f7d53d0eb416f19a91118382b62099f1be2c5b","_number":494509,"virtual_id_number":494509,"owner":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2017-08-18 19:37:34.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2017-08-18 19:29:54.000000000","_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2017-08-18 19:29:54.000000000","_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2017-08-18 19:29:54.000000000","updated_by":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"reviewer":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"state":"REVIEWER"},{"updated":"2017-08-18 19:37:34.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"9b7ece4c35b9ee33d2d88774e32cd708a39efa17","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2017-08-17 12:19:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4cc9a0e7a658e499f54ae24c43a379d8d9ea0f9f","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-17 12:28:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-cookbook-openstack-ops-messaging-chef-rake http://logs.openstack.org/09/494509/1/check/gate-cookbook-openstack-ops-messaging-chef-rake/9b0ac77/ : FAILURE in 2m 59s\n- gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/09/494509/1/check/gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv/22fcd8f/ : FAILURE in 3m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fcb98d98db12e14d21cbc57ad037dae9a13931a2","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2017-08-17 14:27:55.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e39e82d51473a3b3bfd64c0933a9e73c6486b1a5","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-17 17:03:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-cookbook-openstack-ops-messaging-chef-rake http://logs.openstack.org/09/494509/2/check/gate-cookbook-openstack-ops-messaging-chef-rake/6e4b784/ : FAILURE in 2m 17s\n- gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/09/494509/2/check/gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv/941a22a/ : FAILURE in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3dfee284fee1be974576123523892cfaf6239ab4","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2017-08-18 08:07:18.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"f6a9532f55ee9bbe8cbd0a8dc2a98359db8134f8","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 09:47:13.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see http://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n- gate-cookbook-openstack-ops-messaging-chef-rake http://logs.openstack.org/09/494509/3/check/gate-cookbook-openstack-ops-messaging-chef-rake/c59aaca/ : FAILURE in 3m 07s\n- gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/09/494509/3/check/gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv/0624ef5/ : FAILURE in 3m 57s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"aa288a8714b067596d771d855ba86479a0df513f","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-08-18 14:58:43.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"40952385d166bbdbe5273b7b72dcdd9f5857eddf","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 18:19:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cookbook-openstack-ops-messaging-chef-rake http://logs.openstack.org/09/494509/3/check/gate-cookbook-openstack-ops-messaging-chef-rake/27cf24e/ : SUCCESS in 3m 17s\n- gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/09/494509/3/check/gate-cookbook-openstack-ops-messaging-chef-rake-integration-ubuntu-xenial-nv/d373ed1/ : FAILURE in 27m 49s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6bff5d567a65e0ceb0cfdfae25d2c87ee84f7600","author":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"date":"2017-08-18 18:48:44.000000000","message":"Patch Set 3:\n\n* cloudbau-Chef-OpenStack-CI http://ci.cloudbau.de/ci-logs/refs-changes-09-494509-3 : SUCCESS","accounts_in_message":[],"_revision_number":3},{"id":"acce486bff7419db09960490cf1685459f98e4df","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-08-18 19:29:54.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"ec8d52af075fece97993f97945b471d9a181501e","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 19:30:08.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"d40dbd4340cd6c633710c40fe025509d78daec10","date":"2017-08-18 19:37:34.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"b56fd154ea9ff20c189b46b99b6952fe1e27608d","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2017-08-18 19:37:34.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-cookbook-openstack-ops-messaging-chef-rake http://logs.openstack.org/09/494509/3/gate/gate-cookbook-openstack-ops-messaging-chef-rake/7484945/ : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"7318a42ca1a036df29c2b63e11ed00ad5b190d0f","revisions":{"41b8bf65d070bc8057899484c534ba3a5adedd21":{"kind":"REWORK","_number":1,"created":"2017-08-17 12:19:52.000000000","uploader":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"ref":"refs/changes/09/494509/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-ops-messaging","ref":"refs/changes/09/494509/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/1"}}},"commit":{"parents":[{"commit":"67ff49781c2f51255d35fefc04d59ff7945b037b","subject":"add new Chef OpenStack Team Logo to README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/67ff49781c2f51255d35fefc04d59ff7945b037b"}]}],"author":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-17 12:19:46.000000000","tz":120},"committer":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-17 12:19:46.000000000","tz":120},"subject":"starting pike development patch","message":"starting pike development patch\n\nChange-Id: I76c914f68612c0e76dcee8e0557ff62f769533db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/41b8bf65d070bc8057899484c534ba3a5adedd21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/41b8bf65d070bc8057899484c534ba3a5adedd21"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"67ff49781c2f51255d35fefc04d59ff7945b037b","is_merged_in_target_branch":true,"change_id":"Ifd7b8ef493a1810d01a11fa4c1e77dfae7a4ed14","change_number":468807,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"31b8c45e1a46daae2a40cda05f879b33c41efd29":{"kind":"REWORK","_number":2,"created":"2017-08-17 14:27:55.000000000","uploader":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"ref":"refs/changes/09/494509/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-ops-messaging","ref":"refs/changes/09/494509/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/2"}}},"commit":{"parents":[{"commit":"90cf57ae83a52fa2fc3813c6624125ba16401b8b","subject":"Reordered metadata.rb for readability, satisfy foodcritic","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/90cf57ae83a52fa2fc3813c6624125ba16401b8b"}]}],"author":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-17 12:19:46.000000000","tz":120},"committer":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-17 14:27:38.000000000","tz":120},"subject":"starting pike development patch","message":"starting pike development patch\n\nChange-Id: I76c914f68612c0e76dcee8e0557ff62f769533db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/31b8c45e1a46daae2a40cda05f879b33c41efd29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/31b8c45e1a46daae2a40cda05f879b33c41efd29"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"90cf57ae83a52fa2fc3813c6624125ba16401b8b","is_merged_in_target_branch":false,"change_id":"I0dd19692ae9ca1c403762fa19e4c45e60cc76fcd","change_number":494572,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"7318a42ca1a036df29c2b63e11ed00ad5b190d0f":{"kind":"REWORK","_number":3,"created":"2017-08-18 08:07:18.000000000","uploader":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"ref":"refs/changes/09/494509/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-ops-messaging","ref":"refs/changes/09/494509/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/09/494509/3"}}},"commit":{"parents":[{"commit":"59580431cb255e929a4e7e9a4848631d5bbd3a2b","subject":"Reordered metadata.rb for readability, satisfy foodcritic/rubocop","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/59580431cb255e929a4e7e9a4848631d5bbd3a2b"}]}],"author":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-17 12:19:46.000000000","tz":120},"committer":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2017-08-18 08:07:11.000000000","tz":120},"subject":"starting pike development patch","message":"starting pike development patch\n\nChange-Id: I76c914f68612c0e76dcee8e0557ff62f769533db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/7318a42ca1a036df29c2b63e11ed00ad5b190d0f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/7318a42ca1a036df29c2b63e11ed00ad5b190d0f"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"59580431cb255e929a4e7e9a4848631d5bbd3a2b","is_merged_in_target_branch":true,"change_id":"I0dd19692ae9ca1c403762fa19e4c45e60cc76fcd","change_number":494572,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
