)]}'
{"id":"openstack%2Fcookbook-openstack-ops-messaging~523051","triplet_id":"openstack%2Fcookbook-openstack-ops-messaging~master~If971a472c1fd9149a891200548a267655620fce2","project":"openstack/cookbook-openstack-ops-messaging","branch":"master","topic":"bp/modern-chef","hashtags":[],"change_id":"If971a472c1fd9149a891200548a267655620fce2","subject":"ops-messaging refactor for Pike and Chef 13","status":"MERGED","created":"2017-11-27 08:11:31.000000000","updated":"2018-01-17 04:22:29.000000000","submitted":"2018-01-17 04:22:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"523051-1516162949730-35d18e55","meta_rev_id":"e4eeeae98630fb8580fe014ee530b376da7ec306","_number":523051,"virtual_id_number":523051,"owner":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"date":"2018-01-17 03:33:33.000000000","permitted_voting_range":{"min":0,"max":2},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"date":"2018-01-11 19:17:57.000000000","_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":0,"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":2,"date":"2018-01-17 04:22:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2018-01-11 19:50:54.000000000","_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":2,"date":"2018-01-12 09:40:27.000000000","_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":0,"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"value":1,"date":"2018-01-16 13:19:02.000000000","_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2018-01-11 19:50:54.000000000","updated_by":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"reviewer":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"state":"REVIEWER"},{"updated":"2018-01-12 09:40:27.000000000","updated_by":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"reviewer":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"state":"REVIEWER"},{"updated":"2018-01-16 13:19:02.000000000","updated_by":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"reviewer":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"state":"REVIEWER"},{"updated":"2018-01-17 03:33:33.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"REVIEWER"},{"updated":"2018-01-17 04:22:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"09c2a83171d482c08704ab7631b5ff27dbb0ae0b","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-11-27 08:11:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8cd6a0f92710004ca76840e9dd37d695a6815cfc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-27 08:59:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/1/check/openstack-chef-repo-rake/9c2bed7/ : SUCCESS in 2m 18s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/1/check/openstack-chef-repo-integration/3650794/ : FAILURE in 48m 05s","accounts_in_message":[],"_revision_number":1},{"id":"ab1fa16d3cd9eb38b05bacd6f45e261e25e5afb1","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-11-28 04:48:41.000000000","message":"Uploaded patch set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"90dfc67ac588bb56b4a48491dfb596a2f8105bc4","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-28 04:55:22.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/2/check/openstack-chef-repo-rake/e9c5005/ : SUCCESS in 2m 01s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/2/check/openstack-chef-repo-integration/2a8a02e/ : FAILURE in 3m 53s","accounts_in_message":[],"_revision_number":2},{"id":"eb3f9ce1f34c3a60aae0445e1b9082141ad47aca","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-11-28 17:00:50.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"7ec3d09e09f6b47a156886fee68e460644b99ddc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-11-28 17:16:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-rake/9b07b1d/ : SUCCESS in 3m 11s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-integration/06f436a/ : FAILURE in 5m 46s","accounts_in_message":[],"_revision_number":3},{"id":"e03702da1743e2e4d75b7deef876eda409fe52a2","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2017-12-14 17:41:24.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"4790e48aee0412a99e1a9db774ed4b0bcf5256dd","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2017-12-14 17:48:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-rake/d43c010/ : SUCCESS in 3m 00s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-integration/3ea3f54/ : FAILURE in 5m 58s","accounts_in_message":[],"_revision_number":3},{"id":"3444d72338a8c618dba876fea8cb3f16f2b5be6a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2018-01-09 14:07:22.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"afc917a793110f141af7f69d8ef66ca7e3d0c352","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-09 14:25:41.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-rake/0119c91/ : SUCCESS in 3m 58s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-integration/97e6c82/ : FAILURE in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"2630cfe2d09bb644282143cb37446b50da2e950d","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2018-01-11 19:17:57.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"7507716020607a62e6274f169f436221a787e617","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-11 19:46:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-rake/45014aa/ : SUCCESS in 4m 33s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-integration/5dc6325/ : SUCCESS in 26m 51s","accounts_in_message":[],"_revision_number":3},{"id":"d3a6d68a32264ed6d4782a922c89838e8d5a177e","author":{"_account_id":24734,"name":"cloudbau Chef OpenStack CI","email":"ci@cloudbau.de","username":"cloudbau-chef-openstack-ci"},"date":"2018-01-11 19:50:54.000000000","message":"Patch Set 3:\n\n* cloudbau-Chef-OpenStack-CI http://ci.cloudbau.de/ci-logs/refs-changes-51-523051-3 : FAILURE","accounts_in_message":[],"_revision_number":3},{"id":"cc81c42df1c2121fcd28f96c32bf749bb7b9523b","author":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"date":"2018-01-12 09:40:27.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"92de98ec8410491ad851ca175da9ad9ab202cbca","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2018-01-16 13:19:02.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"4dfb505f80916b428480c9c8fb4c45d6fa39b8c0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2018-01-17 03:33:33.000000000","message":"Patch Set 3:\n\nrecheck","accounts_in_message":[],"_revision_number":3},{"id":"14b9df6497f54b810350a9237961ef2f3b9201df","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 03:57:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-rake/0e7941f/ : SUCCESS in 3m 25s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/3/check/openstack-chef-repo-integration/707bd89/ : SUCCESS in 21m 32s","accounts_in_message":[],"_revision_number":3},{"id":"cd92d60bbc9ca229562c12207fc37fd59431cab1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 03:57:20.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"6482e27b60ce9f86fd232797f03053f48e8aed36","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 04:22:29.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"501df45b81e9fa83f14465d1ad3f951e780875ce","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2018-01-17 04:22:29.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-repo-rake http://logs.openstack.org/51/523051/3/gate/openstack-chef-repo-rake/6d78529/ : SUCCESS in 3m 44s\n- openstack-chef-repo-integration http://logs.openstack.org/51/523051/3/gate/openstack-chef-repo-integration/aa573cb/ : SUCCESS in 19m 43s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"d41044ce6489533fb12fb82a9f991d1ae8f1217b","revisions":{"2dbdd83d55118a7123803100753c46fa393c930b":{"kind":"REWORK","_number":1,"created":"2017-11-27 08:11:31.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/51/523051/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-ops-messaging","ref":"refs/changes/51/523051/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/1"}}},"commit":{"parents":[{"commit":"7ddc2b7c7f2ce01b72900003ca1b0c5185cfe215","subject":"Initial ops-messaging Pike updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/7ddc2b7c7f2ce01b72900003ca1b0c5185cfe215"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2017-11-27 08:09:44.000000000","tz":-480},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2017-11-27 08:11:30.000000000","tz":-480},"subject":"ops-messaging refactor for Pike and Chef 13","message":"ops-messaging refactor for Pike and Chef 13\n\n- implemented foodcritic and cookstyle corrections\n- deprecated node.foo.bar method access for node[\u0027foo\u0027][\u0027bar\u0027] bracket access\n- updated for the most recent upstream rabbitmq cookbook\n\nImplements blueprint modern-chef\n\nChange-Id: If971a472c1fd9149a891200548a267655620fce2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/2dbdd83d55118a7123803100753c46fa393c930b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/2dbdd83d55118a7123803100753c46fa393c930b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7ddc2b7c7f2ce01b72900003ca1b0c5185cfe215","is_merged_in_target_branch":true,"change_id":"I84ebeb49823b022e85dddea9fcdd358b27be50a3","change_number":496986,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"6a485030e99cf557125e14cdfa01db0fd39cd500":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2017-11-28 04:48:41.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/51/523051/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-ops-messaging","ref":"refs/changes/51/523051/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/2"}}},"commit":{"parents":[{"commit":"7ddc2b7c7f2ce01b72900003ca1b0c5185cfe215","subject":"Initial ops-messaging Pike updates","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/7ddc2b7c7f2ce01b72900003ca1b0c5185cfe215"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2017-11-27 08:09:44.000000000","tz":-480},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2017-11-28 04:48:27.000000000","tz":-480},"subject":"ops-messaging refactor for Pike and Chef 13","message":"ops-messaging refactor for Pike and Chef 13\n\n- implemented foodcritic and cookstyle corrections\n- deprecated node.foo.bar method access for node[\u0027foo\u0027][\u0027bar\u0027] bracket access\n- updated for the most recent upstream rabbitmq cookbook\n\nImplements blueprint modern-chef\n\nChange-Id: If971a472c1fd9149a891200548a267655620fce2\nDepends-On: I143e0ed0a2bdd76269fc0c402052696426d96d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/6a485030e99cf557125e14cdfa01db0fd39cd500"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/6a485030e99cf557125e14cdfa01db0fd39cd500"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"7ddc2b7c7f2ce01b72900003ca1b0c5185cfe215","is_merged_in_target_branch":true,"change_id":"I84ebeb49823b022e85dddea9fcdd358b27be50a3","change_number":496986,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"d41044ce6489533fb12fb82a9f991d1ae8f1217b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2017-11-28 17:00:50.000000000","uploader":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"ref":"refs/changes/51/523051/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-ops-messaging","ref":"refs/changes/51/523051/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-ops-messaging refs/changes/51/523051/3"}}},"commit":{"parents":[{"commit":"77aad815f7efd074c7aa58235f306c57e9be350b","subject":"Add native zuul v3 jobs defined in openstack-chef-repo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/77aad815f7efd074c7aa58235f306c57e9be350b"}]}],"author":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2017-11-27 08:09:44.000000000","tz":-480},"committer":{"name":"Samuel Cassiba","email":"s@cassiba.com","date":"2017-11-28 17:00:50.000000000","tz":0},"subject":"ops-messaging refactor for Pike and Chef 13","message":"ops-messaging refactor for Pike and Chef 13\n\n- implemented foodcritic and cookstyle corrections\n- deprecated node.foo.bar method access for node[\u0027foo\u0027][\u0027bar\u0027] bracket access\n- updated for the most recent upstream rabbitmq cookbook\n\nImplements blueprint modern-chef\n\nChange-Id: If971a472c1fd9149a891200548a267655620fce2\nDepends-On: I143e0ed0a2bdd76269fc0c402052696426d96d81\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/d41044ce6489533fb12fb82a9f991d1ae8f1217b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-ops-messaging/commit/d41044ce6489533fb12fb82a9f991d1ae8f1217b"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"77aad815f7efd074c7aa58235f306c57e9be350b","is_merged_in_target_branch":true,"change_id":"Ida7a439c71e42359d80b10bff2fb43727866593a","change_number":517260,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
