)]}'
{"id":"openstack%2Fcookbook-openstack-orchestration~411906","triplet_id":"openstack%2Fcookbook-openstack-orchestration~master~Id1fa34e55c556ffae2a6200bb7138f11808a3abf","project":"openstack/cookbook-openstack-orchestration","branch":"master","topic":"newton_deprecations","hashtags":[],"change_id":"Id1fa34e55c556ffae2a6200bb7138f11808a3abf","subject":"RPC_backend / transport_url workover","status":"MERGED","created":"2016-12-16 16:37:23.000000000","updated":"2016-12-21 19:43:10.000000000","submitted":"2016-12-21 19:43:10.000000000","submitter":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"876f86c2e613ea9cc185ac3e03191f19036f466c","_number":411906,"virtual_id_number":411906,"owner":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"all":[{"value":2,"date":"2016-12-21 19:43:10.000000000","post_submit":true,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":0,"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":2,"date":"2016-12-21 19:32:14.000000000","_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":2,"date":"2016-12-20 15:47:38.000000000","_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"all":[{"value":0,"_account_id":3,"name":"Jenkins","username":"jenkins"},{"value":1,"date":"2016-12-21 19:32:14.000000000","_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},{"value":0,"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":3,"name":"Jenkins","username":"jenkins"},{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2016-12-20 15:47:38.000000000","updated_by":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"reviewer":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"state":"REVIEWER"},{"updated":"2016-12-21 19:32:14.000000000","updated_by":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"reviewer":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"state":"REVIEWER"},{"updated":"2016-12-21 19:43:10.000000000","updated_by":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"reviewer":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"state":"REVIEWER"}],"messages":[{"id":"e38e786a9baf2dd6c7007b74fec1b602c2ce8516","author":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"date":"2016-12-16 16:37:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"54df0f23876bed58014a07b2c5e9cd9619a2a56a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-16 16:45:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cookbook-openstack-orchestration-chef-rake http://logs.openstack.org/06/411906/1/check/gate-cookbook-openstack-orchestration-chef-rake/18894bf/ : SUCCESS in 5m 03s\n- gate-cookbook-openstack-orchestration-chef-rake-integration-centos-7-nv http://logs.openstack.org/06/411906/1/check/gate-cookbook-openstack-orchestration-chef-rake-integration-centos-7-nv/30c1db9/ : FAILURE in 3m 35s (non-voting)\n- gate-cookbook-openstack-orchestration-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/06/411906/1/check/gate-cookbook-openstack-orchestration-chef-rake-integration-ubuntu-xenial-nv/66ca098/ : FAILURE in 3m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9e119159b2cdf094d9c5f0fa4d747e8b21644d74","author":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"date":"2016-12-20 08:55:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":1},{"id":"88e1b20f6d33eabf67d536de02a8a3ccb36be10a","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-20 09:15:36.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cookbook-openstack-orchestration-chef-rake http://logs.openstack.org/06/411906/2/check/gate-cookbook-openstack-orchestration-chef-rake/d8f1079/ : SUCCESS in 4m 45s\n- gate-cookbook-openstack-orchestration-chef-rake-integration-centos-7-nv http://logs.openstack.org/06/411906/2/check/gate-cookbook-openstack-orchestration-chef-rake-integration-centos-7-nv/4b3e8c6/ : FAILURE in 7m 30s (non-voting)\n- gate-cookbook-openstack-orchestration-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/06/411906/2/check/gate-cookbook-openstack-orchestration-chef-rake-integration-ubuntu-xenial-nv/21f9171/ : FAILURE in 17m 48s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"b1930d75d3f8164884f9a89f201ad3d2febbb14c","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2016-12-20 11:45:40.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"a5ac24109a9e42ccdac69bb3e8e90a1c9500a004","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-20 12:04:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- gate-cookbook-openstack-orchestration-chef-rake http://logs.openstack.org/06/411906/3/check/gate-cookbook-openstack-orchestration-chef-rake/b500f12/ : SUCCESS in 8m 23s\n- gate-cookbook-openstack-orchestration-chef-rake-integration-centos-7-nv http://logs.openstack.org/06/411906/3/check/gate-cookbook-openstack-orchestration-chef-rake-integration-centos-7-nv/c88f0cf/ : FAILURE in 15m 46s (non-voting)\n- gate-cookbook-openstack-orchestration-chef-rake-integration-ubuntu-xenial-nv http://logs.openstack.org/06/411906/3/check/gate-cookbook-openstack-orchestration-chef-rake-integration-ubuntu-xenial-nv/882e99f/ : FAILURE in 15m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c9b16c5a6a1270538d73e67e6f73c974419e662b","author":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"date":"2016-12-20 15:47:38.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"00f211b1da853d0e48da4b4de5bac38f07dbc1a2","author":{"_account_id":14790,"name":"Samuel Cassiba","email":"samuel@cassi.ba","username":"scassiba"},"date":"2016-12-21 19:32:14.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"c04a2cca131acc361e51e5d1577ec3893f85f1cb","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-21 19:32:25.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.\nhttp://status.openstack.org/zuul/","accounts_in_message":[],"_revision_number":3},{"id":"71a1b785f63d20f449ecef0120529e4387f86edf","date":"2016-12-21 19:43:10.000000000","message":"Change has been successfully merged into the git repository by Jenkins","accounts_in_message":[],"_revision_number":3},{"id":"4acbe2c6f236ce7849268c1f25ddca8fce907bce","author":{"_account_id":3,"name":"Jenkins","username":"jenkins"},"date":"2016-12-21 19:43:10.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- gate-cookbook-openstack-orchestration-chef-rake http://logs.openstack.org/06/411906/3/gate/gate-cookbook-openstack-orchestration-chef-rake/6e0cb86/ : SUCCESS in 10m 10s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"e00cf744d20f501cb0f1172995316ef0489fdc42","revisions":{"b04b387a1f6c4cea5092e060c985caa9054363c3":{"kind":"REWORK","_number":1,"created":"2016-12-16 16:37:23.000000000","uploader":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"ref":"refs/changes/06/411906/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-orchestration","ref":"refs/changes/06/411906/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/1"}}},"commit":{"parents":[{"commit":"a41d278c104a956860ddded8def269c400e1c605","subject":"Show team and repo badges on README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/a41d278c104a956860ddded8def269c400e1c605"}]}],"author":{"name":"Christoph Albers","email":"c.albers@x-ion.de","date":"2016-12-16 16:36:54.000000000","tz":60},"committer":{"name":"Christoph Albers","email":"c.albers@x-ion.de","date":"2016-12-16 16:36:54.000000000","tz":60},"subject":"RPC_backend / transport_url workover","message":"RPC_backend / transport_url workover\n\n- removed deprecated rabbitmq attributes\n- added rabbit_transport_url\n\nChange-Id: Id1fa34e55c556ffae2a6200bb7138f11808a3abf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/b04b387a1f6c4cea5092e060c985caa9054363c3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/b04b387a1f6c4cea5092e060c985caa9054363c3"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"a41d278c104a956860ddded8def269c400e1c605","is_merged_in_target_branch":true,"change_id":"Id903c8625dd3e608cb1db79d40831130a33ad56a","change_number":403347,"patch_set_number":1,"change_status":"MERGED"}],"branch":"refs/heads/master"},"c6abf1a2dded259eff068e0f807665701253fe24":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2016-12-20 08:55:30.000000000","uploader":{"_account_id":19193,"name":"Christoph Albers","email":"c.albers@x-ion.de","username":"calbers"},"ref":"refs/changes/06/411906/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-orchestration","ref":"refs/changes/06/411906/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/2"}}},"commit":{"parents":[{"commit":"b547aaa80c8d1131f3f4a04a6ca4bcb1eef49b98","subject":"Deprecated python-heatclient","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/b547aaa80c8d1131f3f4a04a6ca4bcb1eef49b98"}]}],"author":{"name":"Christoph Albers","email":"c.albers@x-ion.de","date":"2016-12-16 16:36:54.000000000","tz":60},"committer":{"name":"Christoph Albers","email":"c.albers@x-ion.de","date":"2016-12-20 08:55:30.000000000","tz":0},"subject":"RPC_backend / transport_url workover","message":"RPC_backend / transport_url workover\n\n- removed deprecated rabbitmq attributes\n- added rabbit_transport_url\n\nChange-Id: Id1fa34e55c556ffae2a6200bb7138f11808a3abf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/c6abf1a2dded259eff068e0f807665701253fe24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/c6abf1a2dded259eff068e0f807665701253fe24"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"b547aaa80c8d1131f3f4a04a6ca4bcb1eef49b98","is_merged_in_target_branch":true,"change_id":"I54d8190f77b7731dfb518871f080f62718cedc47","change_number":388207,"patch_set_number":2,"change_status":"MERGED"}],"branch":"refs/heads/master"},"e00cf744d20f501cb0f1172995316ef0489fdc42":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2016-12-20 11:45:40.000000000","uploader":{"_account_id":11915,"name":"Jan Klare","email":"klare@osism.tech","username":"jklare"},"ref":"refs/changes/06/411906/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-orchestration","ref":"refs/changes/06/411906/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/06/411906/3"}}},"commit":{"parents":[{"commit":"5943b3b5b3eb1db32dedc285ad8760b175aba23d","subject":"Merge \"Deprecated python-heatclient\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/5943b3b5b3eb1db32dedc285ad8760b175aba23d"}]}],"author":{"name":"Christoph Albers","email":"c.albers@x-ion.de","date":"2016-12-16 16:36:54.000000000","tz":60},"committer":{"name":"Jan Klare","email":"j.klare@cloudbau.de","date":"2016-12-20 11:45:40.000000000","tz":0},"subject":"RPC_backend / transport_url workover","message":"RPC_backend / transport_url workover\n\n- removed deprecated rabbitmq attributes\n- added rabbit_transport_url\n\nChange-Id: Id1fa34e55c556ffae2a6200bb7138f11808a3abf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/e00cf744d20f501cb0f1172995316ef0489fdc42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/e00cf744d20f501cb0f1172995316ef0489fdc42"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"5943b3b5b3eb1db32dedc285ad8760b175aba23d","is_merged_in_target_branch":true}],"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
