)]}'
{"id":"openstack%2Fcookbook-openstack-orchestration~682924","triplet_id":"openstack%2Fcookbook-openstack-orchestration~master~Ic989d9b4c20b6e578e96394dade4e7bef5ae6b58","project":"openstack/cookbook-openstack-orchestration","branch":"master","topic":"chef-py3","hashtags":[],"change_id":"Ic989d9b4c20b6e578e96394dade4e7bef5ae6b58","subject":"Use python3 packages on Ubuntu","status":"MERGED","created":"2019-09-18 14:54:10.000000000","updated":"2019-11-25 15:02:39.000000000","submitted":"2019-11-25 15:02:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"682924-1574694159703-8eeca06d","meta_rev_id":"95534da4fbe3c5acf6acabb43d088ccfc8dcfa77","_number":682924,"virtual_id_number":682924,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-11-25 15:02:39.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"all":[{"value":0,"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":2,"date":"2019-11-25 05:22:50.000000000","_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"all":[{"value":1,"date":"2019-11-25 14:04:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-25 05:22:50.000000000","updated_by":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"reviewer":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"state":"REVIEWER"},{"updated":"2019-11-25 15:02:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"fe5ad78f8a614b7f43943319af5be3703e67016a","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-18 14:54:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"62f544c79239fdb88c15bf5945957b8ab37508c1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-18 15:52:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/da61c91a9023413188159cd9f07848c6 : FAILURE in 4m 41s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/4ee14e57b8b94234b6d678f19a7ed2e7 : SUCCESS in 43m 22s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/5c8c0ac0f5e747cab8eccf69980c71aa : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ed1bab1b5be526fdd8b6862fbddc00c51d8e0ea2","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-09-19 12:15:35.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"cfcd0bb75b88eb0980ce169853bf46d212e69900","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-09-19 13:14:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/5ecf1f63c8924946bdc4169b6bcede45 : SUCCESS in 4m 46s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/de6bd80b567743bdb81fed5d4d94cd9e : SUCCESS in 44m 20s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/81ff50b432604c18bc9f55823c123872 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":2},{"id":"9838290e2c9571e75de4ca3d1f2943cf562d8695","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2019-10-01 21:43:52.000000000","message":"Patch Set 2:\n\nDoes it make sense to finally add this to openstack-chef for proper integration testing?","accounts_in_message":[],"_revision_number":2},{"id":"e70f4087aadfc117d479b6a561fbb1f06b0a91f0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-10-02 08:19:11.000000000","message":"Patch Set 2:\n\nWe had to remove it from the integration test due to memory pressure. See my comments on https://review.opendev.org/#/c/682921/1 for possible solutions.","accounts_in_message":[],"_revision_number":2},{"id":"bf969873e469f13d8d9a945ae816edfb8a957d04","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2019-11-25 05:22:50.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"31208b4cad96fde5254e2af3b03a69dd55070dd1","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-25 11:43:28.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"fa8bdb08d6fd38697445bb929f7eb12b9ed17df8","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-25 14:04:16.000000000","message":"Patch Set 2: -Workflow","accounts_in_message":[],"_revision_number":2},{"id":"bc9461a71c9fa818b8225a644a8d8a63e9c011c9","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2019-11-25 14:04:18.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"017783bddbc5e05030ee7e334de4e15fc381f02d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 14:04:31.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c246279e92e7e56d3cd5a6b6e415d2259b77f37c","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:02:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"ea386af88963476656c1be0d9df8ceb83dad73a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 15:02:39.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/4b2b879a093f41a39d41b1d4ebdc0794 : SUCCESS in 5m 29s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/bebc0c14d8b74de2b6a756e7e287ee3a : SUCCESS in 56m 15s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/9b3ab387152e493188984a918d9394cd : SUCCESS in 48m 35s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"8dfd60f6eca7b6a3ac9f09ee47e6d6457b75a04f","revisions":{"0d6a17325b3e89784645a916cc65a1f9225af542":{"kind":"REWORK","_number":1,"created":"2019-09-18 14:54:10.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/24/682924/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-orchestration","ref":"refs/changes/24/682924/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/1"}}},"commit":{"parents":[{"commit":"cd52c71b6cdb4616c25c12b2bc64a9db84df9dda","subject":"Fixes for fog-openstack, admin endpoint and common pkgs on Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/cd52c71b6cdb4616c25c12b2bc64a9db84df9dda"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-09-18 14:50:12.000000000","tz":0},"committer":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-09-18 14:54:09.000000000","tz":0},"subject":"Use python3 packages on Ubuntu","message":"Use python3 packages on Ubuntu\n\nPython2.7 is going EOL soon, let us deploy python3 for Rocky from the\nstart, so we avoid having to switch later.\n\nChange-Id: Ic989d9b4c20b6e578e96394dade4e7bef5ae6b58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/0d6a17325b3e89784645a916cc65a1f9225af542"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/0d6a17325b3e89784645a916cc65a1f9225af542"}]},"branch":"refs/heads/master"},"8dfd60f6eca7b6a3ac9f09ee47e6d6457b75a04f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2019-09-19 12:15:35.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/24/682924/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstack-orchestration","ref":"refs/changes/24/682924/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstack-orchestration refs/changes/24/682924/2"}}},"commit":{"parents":[{"commit":"cd52c71b6cdb4616c25c12b2bc64a9db84df9dda","subject":"Fixes for fog-openstack, admin endpoint and common pkgs on Debian","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/cd52c71b6cdb4616c25c12b2bc64a9db84df9dda"}]}],"author":{"name":"Jens Harbott","email":"j.harbott@x-ion.de","date":"2019-09-18 14:50:12.000000000","tz":0},"committer":{"name":"Jens Harbott (frickler)","email":"j.harbott@x-ion.de","date":"2019-09-19 12:15:26.000000000","tz":0},"subject":"Use python3 packages on Ubuntu","message":"Use python3 packages on Ubuntu\n\nPython2.7 is going EOL soon, let us deploy python3 for Rocky from the\nstart, so we avoid having to switch later.\n\nDepends-On: https://review.opendev.org/682918\nChange-Id: Ic989d9b4c20b6e578e96394dade4e7bef5ae6b58\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/8dfd60f6eca7b6a3ac9f09ee47e6d6457b75a04f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstack-orchestration/commit/8dfd60f6eca7b6a3ac9f09ee47e6d6457b75a04f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
