)]}'
{"id":"openstack%2Fcookbook-openstackclient~801395","triplet_id":"openstack%2Fcookbook-openstackclient~master~I9c1b8c4e6510b9764b46f2fc8d457b127179d58d","project":"openstack/cookbook-openstackclient","branch":"master","hashtags":[],"change_id":"I9c1b8c4e6510b9764b46f2fc8d457b127179d58d","subject":"Pull bind cookbook from git to fix version pinning issues","status":"MERGED","created":"2021-07-19 22:05:30.000000000","updated":"2021-07-20 00:02:15.000000000","submitted":"2021-07-20 00:02:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"801395","meta_rev_id":"1b1bfa0790ff290b6ebe4a152bec4ab95f1ba675","_number":801395,"virtual_id_number":801395,"owner":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-20 00:02:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"all":[{"value":2,"date":"2021-07-19 23:07:26.000000000","_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"all":[{"value":1,"date":"2021-07-19 23:07:26.000000000","_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-19 22:58:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5d34965ee51a637847f1fac6fef56f3cb524f659","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2021-07-19 22:05:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"19bc66404a6316f85ba5afa29f22e7673b44f3d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 22:58:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/de135a5fe6e44791884a1e56584e4dd6 : SUCCESS in 4m 19s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/534ac05885174708b235367415f76c22 : SUCCESS in 18m 28s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/ad685d12100d437ea1526d70618bfb4d : SUCCESS in 15m 07s\n- openstack-chef-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/3aee173016374afeac0d70734bb3114d : SUCCESS in 48m 09s\n- openstack-chef-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/b6d1189b189f47069eb6a5d68dd9bf22 : SUCCESS in 37m 19s\n- openstack-chef15-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/8c90696deaeb450bb082d3fd315c7603 : SUCCESS in 43m 51s\n- openstack-chef15-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/290a18c1743543eb8bd646ad0691cb94 : SUCCESS in 35m 10s\n- openstack-cinc15-integration-centos https://zuul.opendev.org/t/openstack/build/52f70fa89dda4a24902c9a03c4c30bfb : SUCCESS in 18m 25s\n- openstack-cinc15-integration-ubuntu https://zuul.opendev.org/t/openstack/build/f3f3e8e3403441e59d28b8f55cfedbcf : SUCCESS in 14m 42s\n- openstack-cinc-integration-centos https://zuul.opendev.org/t/openstack/build/dc42cbe476684badb631c16179bb7ca5 : SUCCESS in 17m 42s\n- openstack-cinc-integration-ubuntu https://zuul.opendev.org/t/openstack/build/a63be045faa24be48fa97c0d312fcbb0 : SUCCESS in 13m 41s","accounts_in_message":[],"_revision_number":1},{"id":"11c8e9117054cd19c294f69428202bb3096cecaa","author":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"date":"2021-07-19 23:07:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a6e563d2ebf17acbecfa028687c0959db8ba356","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-19 23:07:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e2c39307130a6ac38befb40c48ed54221a1d74e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:02:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-chef-delivery https://zuul.opendev.org/t/openstack/build/e1212a9639ec49d9a481c15c6f0ee346 : SUCCESS in 4m 27s\n- openstack-chef-integration-centos https://zuul.opendev.org/t/openstack/build/fd6ed9b1ffa94f4eb49415b51c3cc814 : SUCCESS in 19m 07s\n- openstack-chef-integration-ubuntu https://zuul.opendev.org/t/openstack/build/3b147b003a2a4558af08932027c5c5b6 : SUCCESS in 12m 38s\n- openstack-chef-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/a33c33b190b3427cae59b96d03b4b30e : SUCCESS in 42m 23s\n- openstack-chef-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/5cb616f84c5a485a9f754fcda18c6184 : SUCCESS in 33m 58s\n- openstack-chef15-minimal-integration-centos https://zuul.opendev.org/t/openstack/build/e3184f9ba5eb47cdbcc76c833928581c : SUCCESS in 45m 36s\n- openstack-chef15-minimal-integration-ubuntu https://zuul.opendev.org/t/openstack/build/41e023ba284948fca7b87dd0242d3650 : SUCCESS in 43m 28s\n- openstack-cinc15-integration-centos https://zuul.opendev.org/t/openstack/build/2f69c08cb65a4e52988d6d05d2bcea83 : SUCCESS in 19m 13s\n- openstack-cinc15-integration-ubuntu https://zuul.opendev.org/t/openstack/build/47a9cd21c8894dcba80f41d0f0cbd9d9 : SUCCESS in 17m 47s\n- openstack-cinc-integration-centos https://zuul.opendev.org/t/openstack/build/ccfd38e75e0d441faeb95e3ee65dbd35 : SUCCESS in 19m 30s\n- openstack-cinc-integration-ubuntu https://zuul.opendev.org/t/openstack/build/dc671967f59b4600bbbd6dc1a2608735 : SUCCESS in 20m 29s","accounts_in_message":[],"_revision_number":1},{"id":"1b1bfa0790ff290b6ebe4a152bec4ab95f1ba675","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-20 00:02:15.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d159ddb2b73f0a592c2b8c979655e9506a2ec96d","revisions":{"d159ddb2b73f0a592c2b8c979655e9506a2ec96d":{"kind":"REWORK","_number":1,"created":"2021-07-19 22:05:30.000000000","uploader":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"},"ref":"refs/changes/95/801395/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookbook-openstackclient","ref":"refs/changes/95/801395/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookbook-openstackclient refs/changes/95/801395/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookbook-openstackclient refs/changes/95/801395/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookbook-openstackclient refs/changes/95/801395/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookbook-openstackclient refs/changes/95/801395/1"}}},"commit":{"parents":[{"commit":"c7876b15a84e3d177da5d14e96febf49ea5d2f58","subject":"Update to Chef Workstation 21.2.303","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstackclient/commit/c7876b15a84e3d177da5d14e96febf49ea5d2f58"}]}],"author":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2021-07-19 22:05:16.000000000","tz":-420},"committer":{"name":"Lance Albertson","email":"lance@osuosl.org","date":"2021-07-19 22:05:16.000000000","tz":-420},"subject":"Pull bind cookbook from git to fix version pinning issues","message":"Pull bind cookbook from git to fix version pinning issues\n\nChange-Id: I9c1b8c4e6510b9764b46f2fc8d457b127179d58d\nSigned-off-by: Lance Albertson \u003clance@osuosl.org\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstackclient/commit/d159ddb2b73f0a592c2b8c979655e9506a2ec96d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookbook-openstackclient/commit/d159ddb2b73f0a592c2b8c979655e9506a2ec96d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":21961,"name":"Lance Albertson","email":"lance@osuosl.org","username":"ramereth"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
