)]}'
{"id":"openstack%2Fcookiecutter~694312","triplet_id":"openstack%2Fcookiecutter~master~I740c1f7a3ef14c342b3213dca8faeb02e66736bb","project":"openstack/cookiecutter","branch":"master","topic":"update_python_version","hashtags":[],"change_id":"I740c1f7a3ef14c342b3213dca8faeb02e66736bb","subject":"Remove py27 support for future projects","status":"MERGED","created":"2019-11-14 14:33:56.000000000","updated":"2019-12-02 15:33:30.000000000","submitted":"2019-12-02 15:33:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"694312-1575300810377-1e785ae8","meta_rev_id":"b746246919194ab971301271b1ef9f14f9f87cc6","_number":694312,"virtual_id_number":694312,"owner":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-02 15:33:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"all":[{"value":2,"date":"2019-11-14 15:19:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":2,"date":"2019-12-02 15:33:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2019-11-14 15:24:58.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2019-12-02 15:33:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-14 15:19:07.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-11-14 15:24:58.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2019-12-02 15:33:18.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2019-12-02 15:33:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a1148cea6324117288c7a89b3b0ddab8af950758","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-11-14 14:33:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1cb298ed69b4e5c8c452c7d691d4017bc4dc6e58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-14 14:34:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/7a82fb2d90d24ef386db0b832747011c : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"f6ecbcda346a58c30d19e1e616fa22a7be95da4d","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-11-14 15:19:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5e3b24c411498f8afa1ed0df616411f1c9c0fd5f","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-14 15:24:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ccf0b4eee718453072fccab0d3d6b5f6af98619c","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-02 15:33:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(3 comments)\n\nSome follow-ups here but this is a good start","accounts_in_message":[],"_revision_number":1},{"id":"aab822b8b436a8f15d279e486fb42a6da8061e1f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 15:33:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dae984d2a7f0c4b4ca1755d4f7d1a422fdd781c7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 15:33:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"cfdcf48a3221811c048dd79ed266365ef446860b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-02 15:33:30.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/229cd9213a924ce9a8559020512852e2 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cd25bd04ad1cd7003067cd4c125d445ca8cd6f9d","revisions":{"cd25bd04ad1cd7003067cd4c125d445ca8cd6f9d":{"kind":"REWORK","_number":1,"created":"2019-11-14 14:33:56.000000000","uploader":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"ref":"refs/changes/12/694312/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/12/694312/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/12/694312/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/12/694312/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/12/694312/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/12/694312/1"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-11-14 14:33:41.000000000","tz":60},"committer":{"name":"Hervé Beraud","email":"hberaud@redhat.com","date":"2019-11-14 14:33:41.000000000","tz":60},"subject":"Remove py27 support for future projects","message":"Remove py27 support for future projects\n\nChange-Id: I740c1f7a3ef14c342b3213dca8faeb02e66736bb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/cd25bd04ad1cd7003067cd4c125d445ca8cd6f9d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/cd25bd04ad1cd7003067cd4c125d445ca8cd6f9d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
