)]}'
{"id":"openstack%2Fcookiecutter~696001","triplet_id":"openstack%2Fcookiecutter~master~I41381f2ae8a641b8d9052ee8872a460e65c56630","project":"openstack/cookiecutter","branch":"master","topic":"community-goal-template","hashtags":[],"change_id":"I41381f2ae8a641b8d9052ee8872a460e65c56630","subject":"Update CONTRIBUTING.rst template","status":"MERGED","created":"2019-11-26 01:04:47.000000000","updated":"2020-01-17 03:50:20.000000000","submitted":"2020-01-16 19:30:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"696001-1579203059679-5f1f21a0","meta_rev_id":"7671551475ec26a522f31ed7881a67148741699d","_number":696001,"virtual_id_number":696001,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-01-15 19:17:59.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-01-16 19:30:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"recommended":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":1,"date":"2020-01-07 19:14:45.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2020-01-16 19:30:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":2,"date":"2020-01-07 16:05:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"all":[{"value":0,"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":1,"date":"2020-01-16 19:30:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"value":0,"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"value":0,"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},{"value":0,"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-28 14:00:34.000000000","updated_by":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"reviewer":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"state":"REVIEWER"},{"updated":"2019-12-06 10:53:08.000000000","updated_by":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"reviewer":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"state":"REVIEWER"},{"updated":"2019-12-17 15:11:18.000000000","updated_by":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"reviewer":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"state":"REVIEWER"},{"updated":"2019-12-17 19:22:18.000000000","updated_by":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"reviewer":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"state":"REVIEWER"},{"updated":"2019-12-18 20:25:53.000000000","updated_by":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"reviewer":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"state":"REVIEWER"},{"updated":"2020-01-07 16:05:34.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"REVIEWER"},{"updated":"2020-01-07 19:14:45.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-01-16 19:30:47.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-01-16 19:30:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1ce835547b623c1df441f9fc41935e9b2d1f0863","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-11-26 01:04:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d65c7e62f6d29bdd1819143414c3192bee89d18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 01:04:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/f7d754fda5234987a3a01aaf6bb3d96e : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"2a46d61c81be933494e068f5d6a54489274e3587","author":{"_account_id":31245,"name":"Daniel Bengtsson","email":"dbengt@redhat.com","username":"damani42"},"date":"2019-11-28 14:00:34.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"11c47bce897b4ad188ac427f558d84d36d27df09","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-03 21:16:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nThanks, few comments. Are you planning the cookicutter for PTL guide also ?","accounts_in_message":[],"_revision_number":1},{"id":"d8e13a7192adaac728b56c6008a75bebd3d81908","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-05 00:00:33.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"548a8d2a5188bd3075f4cbb9232896ef97871095","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-05 00:00:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1eca7064a730e488330ac6fb2da168cc77b27927","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 00:01:07.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/f2caca3982704f7f9f8b8da8a07105d1 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"f5d01ba2d196323bc0e48a32740985c8ba29e79c","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-05 15:20:26.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"fd6ed13bc3300dd9bc41f24db9873d5a8b7f37eb","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-12-05 15:36:42.000000000","message":"Patch Set 2: Code-Review+1\n\nI think that this approach is good.","accounts_in_message":[],"_revision_number":2},{"id":"1c00f72c350d377177d0985a571d26ae602267df","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-12-05 15:38:59.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nAdded note.","accounts_in_message":[],"_revision_number":2},{"id":"eb8aafdce93f3ae0d71ff8650e96bdde1bdc9ade","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-05 22:27:20.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"119c5d1c08aad4d31e3c26f121b0108fc76170f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 22:27:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/98ae6645acf04d9e851dbe0bc9a914c9 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":3},{"id":"80d0c8d9f5922990e8433a230fb2e4d044d74f06","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-05 22:27:38.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ff54c696dd43fb5ba8bb71867c72860209b94ebf","author":{"_account_id":28522,"name":"Hervé Beraud","email":"herveberaud.pro@gmail.com","username":"hberaud"},"date":"2019-12-06 10:53:08.000000000","message":"Patch Set 3: Code-Review-1\n\n(10 comments)\n\nHello,\n\nFirst, thanks for these changes.\nI\u0027ve Some comments inline (related to -1).\n\nAlso what\u0027s about putting a section related to the developer contribution guide?\n\nhttps://docs.openstack.org/infra/manual/developers.html\n\nIt can be useful to speak about this here.\n\nThoughts?","accounts_in_message":[],"_revision_number":3},{"id":"519a6b0673aa15b400fc6c9304747becccafd151","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-12-06 18:15:40.000000000","message":"Patch Set 3:\n\nThe content looks good.  Just need to address the formatting issues.","accounts_in_message":[],"_revision_number":3},{"id":"996cf194978c3b4adb971c4983a58f8b55ed5fdb","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-09 18:56:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3dd93e3cb3a5ff45c54a6f576ded049b9379935e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-09 18:56:43.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/64c9202b44314fb99f886571ecaa4136 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":4},{"id":"72b51b8053957dab995382f118aae7b0927533f9","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-09 19:12:26.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)\n\nThanks, almost lgtm only one comment about explicit mentioning the project-specific PTL guide link instead of writing those as text in this guide.","accounts_in_message":[],"_revision_number":4},{"id":"fb565157720e7e97240a91807c1822e909d653fd","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-09 21:36:47.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"4fe39b2c94cb5dfee80a4ee269f53fd34bfa9917","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-11 01:26:48.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"eee72bf9735151b6a84c6b83366a67e46ad92ce8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-11 01:27:00.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/f0c84a2152c947faab7ae1c636a0763d : SUCCESS in 0s","accounts_in_message":[],"_revision_number":5},{"id":"64a4472e1910b4a57b66441e0b8bdd7904b0177e","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-12 12:11:41.000000000","message":"Patch Set 5: Code-Review+1\n\nlgtm, thanks","accounts_in_message":[],"_revision_number":5},{"id":"422a98ae752f9cea76cc8a598fc1b6555ef519e3","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-16 23:50:45.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\nProbably should have a patch to the new-project guide too that reminds people to fill out this file, if one hasn\u0027t already been proposed.","accounts_in_message":[],"_revision_number":5},{"id":"2dec658b1b113174b5615dc7e496a2d640e494e3","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-17 02:57:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(6 comments)\n\nSo, uh, this is the CONTRIBUTING guide for the cookiecutter repo... which is an OK place to start I guess... but I\u0027m pretty sure you meant to edit the one for the new repos (cookies?) that get created from this:\n\nhttps://opendev.org/openstack/cookiecutter/src/branch/master/%7b%7bcookiecutter.repo_name%7d%7d/CONTRIBUTING.rst","accounts_in_message":[],"_revision_number":5},{"id":"e8104fdd70058ff3d7037ad4f94427e9c0a9389f","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2019-12-17 13:07:39.000000000","message":"Patch Set 5: Code-Review-1\n\n(3 comments)\n\nWhat Zane said - this is the wrong file :)","accounts_in_message":[],"_revision_number":5},{"id":"32a77f2432c0fbed01fc5f9a2df8d1be33e8684e","author":{"_account_id":6928,"name":"Ben Nemec","email":"openstack@nemebean.com","username":"bnemec"},"date":"2019-12-17 15:11:18.000000000","message":"Patch Set 5: -Code-Review\n\nOh jeez, I completely missed that this is the wrong file.","accounts_in_message":[],"_revision_number":5},{"id":"192c1b7b627bb720e801051591240b554217b437","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2019-12-17 19:22:18.000000000","message":"Patch Set 5: -Code-Review\n\n(1 comment)\n\nremoving my +1 as place for the PTL guide section needs to be figured out first.","accounts_in_message":[],"_revision_number":5},{"id":"05eab505d758cf05af40d9c68473b6587efc4f41","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-17 23:07:16.000000000","message":"Patch Set 5:\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"c37ce6db03a7dad74f10ebd9d684dba2baac462f","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-17 23:09:19.000000000","message":"Patch Set 5:\n\n\u003e (6 comments)\n \u003e \n \u003e So, uh, this is the CONTRIBUTING guide for the cookiecutter repo...\n \u003e which is an OK place to start I guess... but I\u0027m pretty sure you\n \u003e meant to edit the one for the new repos (cookies?) that get created\n \u003e from this:\n \u003e \n \u003e https://opendev.org/openstack/cookiecutter/src/branch/master/%7b%7bcookiecutter.repo_name%7d%7d/CONTRIBUTING.rst\n\nI am happy to duplicate this over there once we have it all settled? Unless it should only live over there?","accounts_in_message":[],"_revision_number":5},{"id":"7adc6d7a1735e6046db70b7bfca8a7f42571684c","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-17 23:11:09.000000000","message":"Patch Set 5:\n\n\u003e \u003e (6 comments)\n \u003e \u003e\n \u003e \u003e So, uh, this is the CONTRIBUTING guide for the cookiecutter\n \u003e repo...\n \u003e \u003e which is an OK place to start I guess... but I\u0027m pretty sure you\n \u003e \u003e meant to edit the one for the new repos (cookies?) that get\n \u003e created\n \u003e \u003e from this:\n \u003e \u003e\n \u003e \u003e https://opendev.org/openstack/cookiecutter/src/branch/master/%7b%7bcookiecutter.repo_name%7d%7d/CONTRIBUTING.rst\n \u003e \n \u003e I am happy to duplicate this over there once we have it all\n \u003e settled? Unless it should only live over there?\n\nOH I think it all just clicked in my brain. My bad.","accounts_in_message":[],"_revision_number":5},{"id":"8232c738193ba096b34473b2f6a03d910deef294","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-17 23:16:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"ff7f564fb5d3562a30f392ebce00b30a753f952e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 23:16:49.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/92accf55020d4ec48aeb6bdb9cc3d989 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":6},{"id":"c491f32c12611a9d348eef9a5120dcee42470562","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-17 23:17:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"984c72d1ed2c36fd9b1221495061a7d203243490","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-17 23:18:10.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/2fc23c065384455b9c3d4fcd502b3ea7 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":7},{"id":"24075aa32af3473bb4c1140e95a1edb26d62c66f","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2019-12-18 05:09:06.000000000","message":"Patch Set 7: Code-Review-1\n\n(3 comments)\n\nActually, on further reflection I\u0027m not convinced that CONTRIBUTING.rst is the best place for this at all. Let\u0027s take a step back and think about what the significance of CONTRIBUTING.rst is: the GitHub UI directs you to it when you attempt to open a pull request for the first time.\n\nSo the #1 message we want to be sending here is DO NOT OPEN A PULL REQUEST. This patch removes that message, that actually appears pretty consistently in every project right now, in favour of a long document that requires a lot of extra reading before somebody would pick up on the fact that PRs will be ignored, and also needs significant customisation by each individual project.\n\nI think the right place for this info (including the PTL stuff) is in the contributor section of the docs published on docs.openstack.org:\n\nhttps://opendev.org/openstack/cookiecutter/src/branch/master/%7b%7bcookiecutter.repo_name%7d%7d/doc/source/contributor/contributing.rst\n\nCONTRIBUTING.rst should probably link to those contributor docs, but I think that\u0027s probably the limit of what we\u0027d want to change. Although I wouldn\u0027t be opposed to repeating the DO NOT OPEN A PULL REQUEST part several times and making it bolder ;)","accounts_in_message":[],"_revision_number":7},{"id":"5b02b65b6b1e2c187bf12e5f636871d40212061f","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-18 18:00:23.000000000","message":"Patch Set 7:\n\n(3 comments)\n\n\u003e (3 comments)\n \u003e \n \u003e Actually, on further reflection I\u0027m not convinced that\n \u003e CONTRIBUTING.rst is the best place for this at all. Let\u0027s take a\n \u003e step back and think about what the significance of CONTRIBUTING.rst\n \u003e is: the GitHub UI directs you to it when you attempt to open a pull\n \u003e request for the first time.\n \u003e \n \u003e So the #1 message we want to be sending here is DO NOT OPEN A PULL\n \u003e REQUEST. This patch removes that message, that actually appears\n \u003e pretty consistently in every project right now, in favour of a long\n \u003e document that requires a lot of extra reading before somebody would\n \u003e pick up on the fact that PRs will be ignored, and also needs\n \u003e significant customisation by each individual project.\n \u003e \n \u003e I think the right place for this info (including the PTL stuff) is\n \u003e in the contributor section of the docs published on\n \u003e docs.openstack.org:\n \u003e \n \u003e https://opendev.org/openstack/cookiecutter/src/branch/master/%7b%7bcookiecutter.repo_name%7d%7d/doc/source/contributor/contributing.rst\n \u003e \n \u003e CONTRIBUTING.rst should probably link to those contributor docs,\n \u003e but I think that\u0027s probably the limit of what we\u0027d want to change.\n \u003e Although I wouldn\u0027t be opposed to repeating the DO NOT OPEN A PULL\n \u003e REQUEST part several times and making it bolder ;)\n\nAh yeah that location makes sense. Will move that in the next patchset.","accounts_in_message":[],"_revision_number":7},{"id":"0c469131e5e35005636d94bffaa04450e407aa8a","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2019-12-18 18:00:31.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"b4cd05bd3a47a84048d248ad35c5be3912c84df5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-18 18:00:44.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/b049719e95174ec6b26f40a963744676 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":8},{"id":"7d1de1cb8b9ed0088bcee58c5cba02e552027ce8","author":{"_account_id":7198,"name":"Jay Bryant","email":"jungleboyj@electronicjungle.net","username":"jsbryant"},"date":"2019-12-18 20:25:53.000000000","message":"Patch Set 8: Code-Review+1\n\nThis looks like it covers everything.","accounts_in_message":[],"_revision_number":8},{"id":"6e1aef72db85167494ac5dcd57fac2ec57c19a27","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2019-12-23 12:46:10.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)\n\nChange itself looks good. -1 is purely for rST issues","accounts_in_message":[],"_revision_number":8},{"id":"91bc7322b82b2037c732a85c840fa1965492a9fd","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-01-07 15:05:27.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"ac15859919ace45a57a00741ca53aec16b811879","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-01-07 15:05:37.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"faa00050ef7c783fcc85d7c63d56efcb333bee7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:06:10.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/dc02f8bff4bf479fbf627ef38f43fe69 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":9},{"id":"fbae3c999a17494683230484293e11960f414364","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-07 15:24:40.000000000","message":"Patch Set 9:\n\n(2 comments)\n\nJust two more spaces to go and we\u0027re golden","accounts_in_message":[],"_revision_number":9},{"id":"94049bf2ec45bc5f114b45ac33dd902dda997259","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-01-07 15:31:05.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"bdf3e52649ce3714bd6fe5737fecd141bef490cf","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-01-07 15:31:05.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"33e29ef457fe0f8b3f55e8fd2c0709002bbd28bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-07 15:31:16.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/bd3f0d957b904937bea760ad4cc349cb : SUCCESS in 0s","accounts_in_message":[],"_revision_number":10},{"id":"c15fd7907e712ce508d8bbc829f38d01735fa230","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2020-01-07 16:05:34.000000000","message":"Patch Set 10: Code-Review+2\n\nLGTM now","accounts_in_message":[],"_revision_number":10},{"id":"dd05c5ec450f8fe47ed82b002aa08d6852967674","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-01-07 19:14:45.000000000","message":"Patch Set 10: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"b9d763dc6b4e5f6010a1c98c8a27f5954d341ac3","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-01-15 19:17:59.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"f9c6f84e1200183dad82c4d574070705304af50d","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-01-16 19:30:47.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"3fca7f7cc10da6c72cede4e441e560de67e9fbdf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:30:58.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"679c06c486fedaefe474435565b5d51cae4b3e47","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:30:59.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/aedd294bb5c64567b3e982f21d6fade3 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":10},{"id":"e098a37909a1e101b3936f8fec52569497d0af32","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-16 19:30:59.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"59d6271521262bdaaeec9590fd70d3a080d7ca15","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-01-17 03:50:20.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2eab7591ef0ed33e563dc7f02a45e6bc2c5eb1d1","revisions":{"89c61d38b7e9c17b099f0ec8db4dc28bd9fad375":{"kind":"REWORK","_number":1,"created":"2019-11-26 01:04:47.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/1"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 01:04:46.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/89c61d38b7e9c17b099f0ec8db4dc28bd9fad375"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/89c61d38b7e9c17b099f0ec8db4dc28bd9fad375"}]},"branch":"refs/heads/master"},"3f2bc58f4ceee1e055808befc2b14d93772769c0":{"kind":"REWORK","_number":2,"created":"2019-12-05 00:00:56.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/2"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-12-05 00:00:46.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/3f2bc58f4ceee1e055808befc2b14d93772769c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/3f2bc58f4ceee1e055808befc2b14d93772769c0"}]},"branch":"refs/heads/master"},"02fc1467b85a550c84d3bb8363a9795fd47a4b10":{"kind":"REWORK","_number":3,"created":"2019-12-05 22:27:20.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/3"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-12-05 22:27:10.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/02fc1467b85a550c84d3bb8363a9795fd47a4b10"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/02fc1467b85a550c84d3bb8363a9795fd47a4b10"}]},"branch":"refs/heads/master"},"093d5ac4dd93403c49b3690da201bd838104d059":{"kind":"REWORK","_number":4,"created":"2019-12-09 18:56:30.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/4"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-12-09 18:56:22.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/093d5ac4dd93403c49b3690da201bd838104d059"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/093d5ac4dd93403c49b3690da201bd838104d059"}]},"branch":"refs/heads/master"},"0fe75da2abb1d19377b96ca6f92bb7154c975f05":{"kind":"REWORK","_number":5,"created":"2019-12-11 01:26:48.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/5"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-12-11 01:26:40.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/0fe75da2abb1d19377b96ca6f92bb7154c975f05"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/0fe75da2abb1d19377b96ca6f92bb7154c975f05"}]},"branch":"refs/heads/master"},"b6a13691c74e644d974fb7f8a5c3dcbc4004f637":{"kind":"REWORK","_number":6,"created":"2019-12-17 23:16:38.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/6"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-12-17 23:15:50.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b6a13691c74e644d974fb7f8a5c3dcbc4004f637"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b6a13691c74e644d974fb7f8a5c3dcbc4004f637"}]},"branch":"refs/heads/master"},"8221847eab643aa81e1f2efea4e94b6e621d6947":{"kind":"REWORK","_number":7,"created":"2019-12-17 23:17:59.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/7"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-12-17 23:17:50.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/8221847eab643aa81e1f2efea4e94b6e621d6947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/8221847eab643aa81e1f2efea4e94b6e621d6947"}]},"branch":"refs/heads/master"},"a010e9859ace588be0551f2408832a943b63fff8":{"kind":"REWORK","_number":8,"created":"2019-12-18 18:00:31.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/8"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-12-18 17:59:35.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/a010e9859ace588be0551f2408832a943b63fff8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/a010e9859ace588be0551f2408832a943b63fff8"}]},"branch":"refs/heads/master"},"867e3b618b3a324be5f19c1b159d560c53f676f1":{"kind":"REWORK","_number":9,"created":"2020-01-07 15:05:37.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/9"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-01-07 15:05:08.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/867e3b618b3a324be5f19c1b159d560c53f676f1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/867e3b618b3a324be5f19c1b159d560c53f676f1"}]},"branch":"refs/heads/master"},"2eab7591ef0ed33e563dc7f02a45e6bc2c5eb1d1":{"kind":"REWORK","_number":10,"created":"2020-01-07 15:31:05.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/01/696001/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/01/696001/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/01/696001/10"}}},"commit":{"parents":[{"commit":"b67d3d35c7d3068b76bb997e4ac14597bfda6b8b","subject":"Update the python version in envlist attribut.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/b67d3d35c7d3068b76bb997e4ac14597bfda6b8b"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2019-11-26 00:56:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-01-07 15:30:55.000000000","tz":-480},"subject":"Update CONTRIBUTING.rst template","message":"Update CONTRIBUTING.rst template\n\nAs a part of the Ussuri Community Goal, the template here needs to\nexpanded to outline the things that are consistenly different between\nprojects or different from the norm that should be documented so that\nnew contributors can more easy onboard themselves to a given project.\n\nGoal Patch: https://review.opendev.org/#/c/691737/2\n\nChange-Id: I41381f2ae8a641b8d9052ee8872a460e65c56630\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/2eab7591ef0ed33e563dc7f02a45e6bc2c5eb1d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/2eab7591ef0ed33e563dc7f02a45e6bc2c5eb1d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
