)]}'
{"id":"openstack%2Fcookiecutter~707735","triplet_id":"openstack%2Fcookiecutter~master~Idadd0a0f1fda536bf5ad31d0c4164a45baf2c68b","project":"openstack/cookiecutter","branch":"master","topic":"add_include","hashtags":[],"change_id":"Idadd0a0f1fda536bf5ad31d0c4164a45baf2c68b","subject":"Add include for CONTRIBUTING.rst","status":"ABANDONED","created":"2020-02-13 22:41:39.000000000","updated":"2020-02-18 20:44:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"44c0c2cfc650eb5c605d39183f9b9ea64e711752","_number":707735,"virtual_id_number":707735,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-02-18 17:42:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2020-02-18 19:29:20.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-17 16:15:47.000000000","updated_by":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"reviewer":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"state":"REVIEWER"},{"updated":"2020-02-18 04:58:10.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-02-18 17:42:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-02-18 19:29:20.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"}],"messages":[{"id":"58b45049fc9398ad044e34247119210e128cf92b","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-02-13 22:41:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"de6ea0c2be4bac241c83b3257103fb77d9b0fbad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-13 22:41:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/958058cee2024e7fb751bc7adbc675dd : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"16a223107c176800a0daacf836be5c6a227f6e1d","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-14 08:46:50.000000000","message":"Patch Set 1:\n\nWouldn\u0027t it be better to have the include the other way around? In this way, CONTRIBUTING.rst contains only the notice about not accepting PRs in Github, but no references to the place where the documentation lives.\n\nI would add an explicit suggestion about stripping CONTRIBUTING.rst from any content but the \"Do not accept...\" and a pointer to the real documentation (see here https://docs.... contributing.html for more details).\n-","accounts_in_message":[],"_revision_number":1},{"id":"53a8fbc3b7195e0df823855119d33a8a4eaba981","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-02-17 14:47:48.000000000","message":"Patch Set 1:\n\nIMHO this is not really necessary. The info in CONTRIBUTING.rst is pretty narrow and doesn\u0027t add a lot to what is on this page, which already links to the OpenStack contributor guide and the bug trackers.\n\nWe cant do this the other way round because the include has to be processed by Sphinx to work. But now that we have this contributing page in a common location we could possibly create a link to this doc from CONTRIBUTING.rst; that would probably be valuable.","accounts_in_message":[],"_revision_number":1},{"id":"ec80187b636c00d6cf5e22dc2e10a192f88777db","author":{"_account_id":10459,"name":"Luigi Toscano","email":"ltoscano@redhat.com","username":"ltoscano"},"date":"2020-02-17 16:15:47.000000000","message":"Patch Set 1:\n\n\u003e IMHO this is not really necessary. The info in CONTRIBUTING.rst is\n \u003e pretty narrow and doesn\u0027t add a lot to what is on this page, which\n \u003e already links to the OpenStack contributor guide and the bug\n \u003e trackers.\n\nIf the info in CONTRIBUTING.rst is pretty narrow, why do we need to include it? Just let it say that you should not use github for PR (which is the main goal) and refer to the proper documentation on https://docs.openstack.org/....contributing.html.\n\n \u003e We cant do this the other way round because the include has to be\n \u003e processed by Sphinx to work. But now that we have this contributing\n \u003e page in a common location we could possibly create a link to this\n \u003e doc from CONTRIBUTING.rst; that would probably be valuable.\n\nThis is what I suggested, in fact: explicitly reference the published contributing page  from CONTRIBUTING.rst. I just don\u0027t see the reason for including the top-level CONTRIBUTING.rst.","accounts_in_message":[],"_revision_number":1},{"id":"65def2734572649e929867c10d7d53943cb74a00","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-02-18 04:58:10.000000000","message":"Patch Set 1:\n\nI think we\u0027re in violent agreement :)","accounts_in_message":[],"_revision_number":1},{"id":"fb4f455a701817ca054fb7b59805d21f799a1ed7","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-02-18 17:38:53.000000000","message":"Patch Set 1:\n\n\u003e \u003e IMHO this is not really necessary. The info in CONTRIBUTING.rst\n \u003e is\n \u003e \u003e pretty narrow and doesn\u0027t add a lot to what is on this page,\n \u003e which\n \u003e \u003e already links to the OpenStack contributor guide and the bug\n \u003e \u003e trackers.\n \u003e \n \u003e If the info in CONTRIBUTING.rst is pretty narrow, why do we need to\n \u003e include it? Just let it say that you should not use github for PR\n \u003e (which is the main goal) and refer to the proper documentation on\n \u003e https://docs.openstack.org/....contributing.html.\n \u003e \n \u003e \u003e We cant do this the other way round because the include has to be\n \u003e \u003e processed by Sphinx to work. But now that we have this\n \u003e contributing\n \u003e \u003e page in a common location we could possibly create a link to this\n \u003e \u003e doc from CONTRIBUTING.rst; that would probably be valuable.\n \u003e \n \u003e This is what I suggested, in fact: explicitly reference the\n \u003e published contributing page  from CONTRIBUTING.rst. I just don\u0027t\n \u003e see the reason for including the top-level CONTRIBUTING.rst.\n\nThat\u0027s definitely what I meant to do, but got it the wrong way round in my head. Fixing in the next patchset.","accounts_in_message":[],"_revision_number":1},{"id":"708673d7d95334c7727f9b031f52cc5bdbb87460","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-02-18 17:42:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c4bbb7df2f9678e97339d612aab4fc6e9fb2e62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 17:42:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/4f1e1b12be184624aa0c84096c7252b1 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"75ed7bc94b9f878ecb2ecbe3b15c6c2c9cdae7ba","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-18 19:29:20.000000000","message":"Patch Set 2: Code-Review-1\n\nI really think this should be the other way around. When someone clones the repo they ought to be able to less or cat CONTRIBUTING.rst and find out the information they need without being directed to go looking in some Sphinx source file several directories deep. On the other hand, Sphinx will do an excellent job of parsing an include directive in the documentation source and pulling in content from the CONTRIBUTING.rst file.","accounts_in_message":[],"_revision_number":2},{"id":"5fa3069e04d5fb319c8f656d6bbdb43d452a3185","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-18 20:15:31.000000000","message":"Patch Set 2:\n\nOh, also this is changing the CONTRIBUTING.rst for the cookiecutter project, not the {{cookiecutter.repo_name}}/CONTRIBUTING.rst template, which I think is what you meant to change based on your commit message.","accounts_in_message":[],"_revision_number":2},{"id":"903de02fca7d833e9e73b7f66c30c4f880995d4a","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-18 20:27:27.000000000","message":"Patch Set 2:\n\nIt looks like I41381f2ae8a641b8d9052ee8872a460e65c56630 discussion happened while I was away from Internet access, and ended up with a split of contributor information between {{cookiecutter.repo_name}}/CONTRIBUTING.rst and {{cookiecutter.repo_name}}/doc/source/contributor/contributing.rst rather than keeping all content in the former and preserving the latter as only a shell with an include directive. While I think that OpenStack projects ought to be able to do either and satisfy the cycle goal, I\u0027d prefer if we didn\u0027t recommend such a complex split in the cookiecutter template.","accounts_in_message":[],"_revision_number":2},{"id":"0ac3549f2fafa34bb8b11982e4ebf8caac9940dc","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-02-18 20:44:39.000000000","message":"Abandoned\n\nAbandoning this in favor of: https://review.opendev.org/708511","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"82ea9c829fe2d5c8abbb101535f14e75831fb613","revisions":{"f20ae4e829dcbf23851386312833e0a7fce39b23":{"kind":"REWORK","_number":1,"created":"2020-02-13 22:41:39.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/35/707735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/35/707735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/1"}}},"commit":{"parents":[{"commit":"4fbd387ddac45dc56303e26ea6eb4e42da6bc2f0","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/4fbd387ddac45dc56303e26ea6eb4e42da6bc2f0"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-13 22:35:57.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-13 22:35:57.000000000","tz":-480},"subject":"Add include for CONTRIBUTING.rst","message":"Add include for CONTRIBUTING.rst\n\nOne of the most important things people need to see when they\nlook at a repo is that we DO NOT ACCEPT PRs IN GITHUB. This\ninclude adds that info which exists in CONTRIBUTING.rst.\n\nChange-Id: Idadd0a0f1fda536bf5ad31d0c4164a45baf2c68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/f20ae4e829dcbf23851386312833e0a7fce39b23"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/f20ae4e829dcbf23851386312833e0a7fce39b23"}]},"branch":"refs/heads/master"},"82ea9c829fe2d5c8abbb101535f14e75831fb613":{"kind":"REWORK","_number":2,"created":"2020-02-18 17:42:37.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/35/707735/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/35/707735/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/35/707735/2"}}},"commit":{"parents":[{"commit":"4fbd387ddac45dc56303e26ea6eb4e42da6bc2f0","subject":"[ussuri][goal] Drop python 2.7 support and testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/4fbd387ddac45dc56303e26ea6eb4e42da6bc2f0"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-13 22:35:57.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-18 17:42:25.000000000","tz":-480},"subject":"Add include for CONTRIBUTING.rst","message":"Add include for CONTRIBUTING.rst\n\nOne of the most important things people need to see when they\nlook at a repo is that we DO NOT ACCEPT PRs IN GITHUB. This\ninclude adds that info which exists in CONTRIBUTING.rst.\n\nChange-Id: Idadd0a0f1fda536bf5ad31d0c4164a45baf2c68b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/82ea9c829fe2d5c8abbb101535f14e75831fb613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/82ea9c829fe2d5c8abbb101535f14e75831fb613"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
