)]}'
{"id":"openstack%2Fcookiecutter~708511","triplet_id":"openstack%2Fcookiecutter~master~I2bcda1e0f14dac7843869666df4250604b64c265","project":"openstack/cookiecutter","branch":"master","topic":"template_location","hashtags":[],"change_id":"I2bcda1e0f14dac7843869666df4250604b64c265","subject":"Fix Contributing template location","status":"MERGED","created":"2020-02-18 20:40:09.000000000","updated":"2020-02-19 17:22:33.000000000","submitted":"2020-02-19 15:39:05.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"708511-1582126745753-a5ae4dff","meta_rev_id":"52b310923a129108a7554ff6621f6ad4408677c4","_number":708511,"virtual_id_number":708511,"owner":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"date":"2020-02-19 15:36:04.000000000","_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"date":"2020-02-18 20:56:16.000000000","_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-19 15:39:05.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":2,"date":"2020-02-19 15:38:54.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":2,"date":"2020-02-18 22:47:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2020-02-18 21:21:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"all":[{"value":1,"date":"2020-02-19 15:38:54.000000000","_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"value":0,"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"value":0,"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"value":0,"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-02-18 21:21:34.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"REVIEWER"},{"updated":"2020-02-18 22:47:00.000000000","updated_by":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"reviewer":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"state":"REVIEWER"},{"updated":"2020-02-19 15:36:04.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"REVIEWER"},{"updated":"2020-02-19 15:38:54.000000000","updated_by":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"reviewer":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"state":"REVIEWER"},{"updated":"2020-02-19 15:39:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3366090c63c57339540a8ef47b0a77e8184e94db","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-02-18 20:40:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b8187aa8857e5c457ea7c5564368fa5d31ca7e5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:40:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/cd98e02ec6804a40994e4bb59b784022 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":1},{"id":"8e0f6c5c8e53ea918cf8d4d96efa4ec432324058","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-18 20:44:03.000000000","message":"Patch Set 1: Code-Review+1\n\nNot just GitHub, but more generally when someone clones the repo and looks at the top-level files like the readme and license, they also have a file detailing how to contribute changes back to the maintainers without having to go digging several levels deep into the source tree for Sphinx-built docs.","accounts_in_message":[],"_revision_number":1},{"id":"665a5522fa5ed92827867c4e1fc6e1993d42707e","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-18 20:44:45.000000000","message":"Patch Set 1: Code-Review-1\n\nOh, except that you wanted to move it into {{cookiecutter.repo_name}}/CONTRIBUTING.rst I think.","accounts_in_message":[],"_revision_number":1},{"id":"352107b296270c2ba1163314d9a7f7e40d4c6d16","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-18 20:48:52.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a250be401b6c3353cc4d6a2663dd43b59fd19623","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-02-18 20:53:32.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"2f1ff37fef2bee561c141444485d5135efdfd2d9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-18 20:53:44.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/b8f9fe0204244c62b558c4ed7e185f3f : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"7e63fbf4cad6b2e94fc5eda7ff4daba86df0deef","author":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"date":"2020-02-18 20:56:16.000000000","message":"Patch Set 2:\n\n\u003e (1 comment)\n\nThat top level CONTRIBUTING keeps confusing me. Should be fine now.","accounts_in_message":[],"_revision_number":2},{"id":"1084629276f310c4bbc77ef023dc545a286fec45","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2020-02-18 21:21:34.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"8442c31556c7b26288e8adc244d9dc49d9e9634c","author":{"_account_id":2,"name":"Monty Taylor","email":"mordred@inaugust.com","username":"mordred"},"date":"2020-02-18 22:47:00.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"6a15e1d0bffbc66fc91a7ff624978407a96c3907","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-02-19 15:23:05.000000000","message":"Patch Set 2:\n\nI really don\u0027t agree with Jeremy\u0027s feedback on the other patch that led to this. If CONTRIBUTING.rst contains anything more complicated than \"Don\u0027t use GitHub; here are the contributor docs\" then we will have a plague of people trying to use GitHub, because people don\u0027t read.\n\n\u003e more generally when someone clones the repo and looks at the top-level files like the readme and license, they also have a file detailing how to contribute changes back to the maintainers without having to go digging several levels deep into the source tree for Sphinx-built docs.\n\nI agree; nobody thinks that you should have to go digging through the source.\n\nBut a link to docs.o.o should be fine IMHO.\n\nNot gonna -1 because it must be hard enough refereeing this argument already :(","accounts_in_message":[],"_revision_number":2},{"id":"dc8750ab091726e74bc14290083683e3c5876164","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2020-02-19 15:36:04.000000000","message":"Patch Set 2:\n\nFor clarity, this is what I\u0027m suggesting: I02dd34a1cf6b62d12f7a26d21b3a2857d82d1b0a","accounts_in_message":[],"_revision_number":2},{"id":"dfb68f122210275c36cf16a45b4f2ee3d6caae51","author":{"_account_id":11904,"name":"Sean McGinnis","email":"sean.mcginnis@gmail.com","username":"SeanM"},"date":"2020-02-19 15:38:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)\n\nMinor nit that can be addressed in a follow up.","accounts_in_message":[],"_revision_number":2},{"id":"865ec4ec61a6dc3775bc817546869adbf5897dee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 15:39:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5e97f05a7a2042c260a7d6db03d27529f04edacc","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 15:39:05.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"1f48e7c7dbe5baea26abdd17121453ec76939c9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-19 15:39:05.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- noop https://zuul.opendev.org/t/openstack/build/ae86ed4d3a5541018f8e25dbf9be6d98 : SUCCESS in 0s","accounts_in_message":[],"_revision_number":2},{"id":"3e6cb8d4fc4974dbc2ed143934999b0648b67384","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2020-02-19 17:22:33.000000000","message":"Patch Set 2:\n\n\u003e I really don\u0027t agree with Jeremy\u0027s feedback on the other patch that\n \u003e led to this. If CONTRIBUTING.rst contains anything more complicated\n \u003e than \"Don\u0027t use GitHub; here are the contributor docs\" then we will\n \u003e have a plague of people trying to use GitHub, because people don\u0027t\n \u003e read.\n \u003e \n \u003e \u003e more generally when someone clones the repo and looks at the\n \u003e top-level files like the readme and license, they also have a file\n \u003e detailing how to contribute changes back to the maintainers without\n \u003e having to go digging several levels deep into the source tree for\n \u003e Sphinx-built docs.\n \u003e \n \u003e I agree; nobody thinks that you should have to go digging through\n \u003e the source.\n \u003e \n \u003e But a link to docs.o.o should be fine IMHO.\n \u003e \n \u003e Not gonna -1 because it must be hard enough refereeing this\n \u003e argument already :(\n\nI am ok with having all info in CONTRIBUTING.rst but we can highlight the first two line in bold so that people at least read it when open this file. rest all detailed info go into next para and so\n\n----------------\nOPENSTACK USES GERRIT FOR CODE REVIEW.\n\nANY PULL REQUESTS SUBMITTED THROUGH GITHUB WILL BE IGNORED.\n------------","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ed3a0b07c72667628064dc7c5d8eae47dcae16d6","revisions":{"201db82a4d63c168490416f8c966fc57f95e5c37":{"kind":"REWORK","_number":1,"created":"2020-02-18 20:40:09.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/11/708511/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/11/708511/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/1"}}},"commit":{"parents":[{"commit":"c96fee604917fbe896380e99c3219e55dc405f9a","subject":"Fix gitreview host name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/c96fee604917fbe896380e99c3219e55dc405f9a"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-18 20:34:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-18 20:40:08.000000000","tz":-480},"subject":"Fix Contributing template location","message":"Fix Contributing template location\n\nThere were several discussions[1][2] on includes and\nwhere the template should actually be living for the U\ncommunity goal about contributor docs.\n\nHaving the information in CONTRIBUTING.rst so that it\nappears on github and using an include in\ndoc/source/contributor/contributing.rst to pull that text\nin via Sphinx is a cleaner way of keeping all the info\ntogether.\n\n[1] https://review.opendev.org/696001\n[2] https://review.opendev.org/#/c/707735/\n\nChange-Id: I2bcda1e0f14dac7843869666df4250604b64c265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/201db82a4d63c168490416f8c966fc57f95e5c37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/201db82a4d63c168490416f8c966fc57f95e5c37"}]},"branch":"refs/heads/master"},"ed3a0b07c72667628064dc7c5d8eae47dcae16d6":{"kind":"REWORK","_number":2,"created":"2020-02-18 20:53:32.000000000","uploader":{"_account_id":16708,"name":"Kendall Nelson","display_name":"Kendall (diablo_rojo)","email":"kennelson11@gmail.com","username":"kjnelson"},"ref":"refs/changes/11/708511/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cookiecutter","ref":"refs/changes/11/708511/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cookiecutter refs/changes/11/708511/2"}}},"commit":{"parents":[{"commit":"c96fee604917fbe896380e99c3219e55dc405f9a","subject":"Fix gitreview host name","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/c96fee604917fbe896380e99c3219e55dc405f9a"}]}],"author":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-18 20:34:55.000000000","tz":-480},"committer":{"name":"Kendall Nelson","email":"kennelson11@gmail.com","date":"2020-02-18 20:53:22.000000000","tz":-480},"subject":"Fix Contributing template location","message":"Fix Contributing template location\n\nThere were several discussions[1][2] on includes and\nwhere the template should actually be living for the U\ncommunity goal about contributor docs.\n\nHaving the information in CONTRIBUTING.rst so that it\nappears on github and using an include in\ndoc/source/contributor/contributing.rst to pull that text\nin via Sphinx is a cleaner way of keeping all the info\ntogether.\n\n[1] https://review.opendev.org/696001\n[2] https://review.opendev.org/#/c/707735/\n\nChange-Id: I2bcda1e0f14dac7843869666df4250604b64c265\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/ed3a0b07c72667628064dc7c5d8eae47dcae16d6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cookiecutter/commit/ed3a0b07c72667628064dc7c5d8eae47dcae16d6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
