)]}'
{"id":"openstack%2Fcyborg-specs~658263","triplet_id":"openstack%2Fcyborg-specs~master~I735b296aab1037efdd16eb99d5b5e5da56f95746","project":"openstack/cyborg-specs","branch":"master","topic":"658263","hashtags":[],"change_id":"I735b296aab1037efdd16eb99d5b5e5da56f95746","subject":"Specification for Cyborg APIs Version 2.","status":"MERGED","created":"2019-05-10 04:50:51.000000000","updated":"2019-07-24 03:14:11.000000000","submitted":"2019-07-24 03:14:11.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":33,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"658263-1563938051868-15be325b","meta_rev_id":"f343d6ffcd479fb33ee54077e513d304ba8ca444","_number":658263,"virtual_id_number":658263,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2019-07-24 03:14:11.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":2,"date":"2019-07-24 03:08:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-07-16 05:29:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":0,"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":1,"date":"2019-07-24 03:07:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 23:18:15.000000000","updated_by":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"reviewer":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"state":"REVIEWER"},{"updated":"2019-05-10 23:18:23.000000000","updated_by":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"reviewer":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"state":"REVIEWER"},{"updated":"2019-05-10 23:19:23.000000000","updated_by":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"reviewer":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"state":"REVIEWER"},{"updated":"2019-05-17 06:28:27.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"},{"updated":"2019-05-17 07:39:48.000000000","updated_by":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"reviewer":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"state":"REVIEWER"},{"updated":"2019-07-03 03:52:17.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-07-05 08:06:10.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-07-16 05:29:39.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2019-07-24 03:07:16.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-07-24 03:14:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"f81bbe6213ba0e207ded365b0a776bc29017d286","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-10 04:50:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3b07280a0ea7eee1894d413a680685be886822b3","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-10 04:54:44.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"d86e447030db789f18eaac06ed7fb46964ea72d1","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 04:59:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/658263/1/check/openstack-tox-docs/2901c27/html/ : SUCCESS in 3m 35s\n- openstack-tox-pep8 http://logs.openstack.org/63/658263/1/check/openstack-tox-pep8/644680d/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":1},{"id":"fce95744d2f77b693e2a0ab0394a2ebb79a504f2","author":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"date":"2019-05-17 06:28:27.000000000","message":"Patch Set 1: Code-Review-1\n\n(5 comments)\n\nPlease see my comments in line.","accounts_in_message":[],"_revision_number":1},{"id":"3fb60505e3db9b066ce8a2e161e2480bbf95c6d5","author":{"_account_id":29845,"name":"Ikuo Otani","email":"ikuo.otani.rw@gmail.com","username":"ikuo_o"},"date":"2019-05-17 07:39:48.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nPlease see my questions about APIs in line.","accounts_in_message":[],"_revision_number":1},{"id":"5baa88915d6752fe6cdb80512841cfb3c33660cd","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-06 03:59:11.000000000","message":"Patch Set 1:\n\n(7 comments)\n\nThanks, Li and Ikuo.","accounts_in_message":[],"_revision_number":1},{"id":"695d4bf9c94c5bd302f0431424262ca23089671e","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-06-20 02:13:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"06f0e54b8359fa3dc779565623e361a4809e750d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-28 01:10:12.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nThanks for the comments.","accounts_in_message":[],"_revision_number":1},{"id":"c54fdf6410db18a6244f580341e51c36e559290d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-28 01:29:05.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.","accounts_in_message":[],"_revision_number":2},{"id":"4509e37782020950cabbd402dbcbf9256d13e69d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-28 01:32:50.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c90f8b158e99e9ddb873b17bd46c79c47ec62be0","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-06-28 01:44:43.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/658263/3/check/openstack-tox-docs/9ddfcb5/html/ : SUCCESS in 2m 53s\n- openstack-tox-pep8 http://logs.openstack.org/63/658263/3/check/openstack-tox-pep8/a98da55/ : SUCCESS in 3m 08s","accounts_in_message":[],"_revision_number":3},{"id":"8fca477c6166306a8e32cfeb08075b8fd1128ffb","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-03 03:52:17.000000000","message":"Patch Set 3: Code-Review+1\n\n(5 comments)\n\nLGTM, and some nits see comments inline.","accounts_in_message":[],"_revision_number":3},{"id":"fd0ed208802ce7f9e490dc629f7d4e14294a868c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-03 21:30:11.000000000","message":"Patch Set 3:\n\n(6 comments)\n\nThanks, Yikun.","accounts_in_message":[],"_revision_number":3},{"id":"805191e1d8d3a10ecd345d1a57766b01843d1b4a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-03 21:31:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"56651a1dec292d1a8a6c31e3929cd0ffb3d429bf","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-03 21:36:57.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/658263/4/check/openstack-tox-docs/57dceb6/html/ : SUCCESS in 3m 38s\n- openstack-tox-pep8 http://logs.openstack.org/63/658263/4/check/openstack-tox-pep8/2e3250e/ : SUCCESS in 3m 49s","accounts_in_message":[],"_revision_number":4},{"id":"27f1beb7d079185063af36d774d435b7b3342eef","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-07-05 08:06:10.000000000","message":"Patch Set 4: Code-Review+1\n\nLGTM.","accounts_in_message":[],"_revision_number":4},{"id":"815c3788ba00ca0bc89eb92154121754c7101050","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-07-09 11:37:18.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"cd3ee469051239c5d0f1b7a907762a599dee8f5c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 08:41:25.000000000","message":"Patch Set 4:\n\n(1 comment)\n\nThanks, Yumeng.","accounts_in_message":[],"_revision_number":4},{"id":"f71c34a63b0e1083f2d3760cfc058d37c5de0154","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 08:41:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"4c6f3d79ca32cc0dcae7a58abd40b253751c3de4","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-12 08:44:02.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f97a30736920acf089bec2770ce616d9dc168d27","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-12 08:52:25.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/658263/6/check/openstack-tox-docs/7db372b/html/ : SUCCESS in 2m 36s\n- openstack-tox-pep8 http://logs.openstack.org/63/658263/6/check/openstack-tox-pep8/80b8464/ : SUCCESS in 2m 22s","accounts_in_message":[],"_revision_number":6},{"id":"fdef46b84ab1493c78643372e70cc27ac02aa65c","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-07-16 05:29:39.000000000","message":"Patch Set 6: Code-Review+2\n\nLGTM.\nI created [0] so contributors interested in DB can pick.\n[0] https://storyboard.openstack.org/#!/story/2006226","accounts_in_message":[],"_revision_number":6},{"id":"017f05834eeaeea0c4705c4c01e91444deb293d1","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-07-24 03:07:16.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"102e34fde47485c29ead0e4409c0ac103f601b37","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 03:07:27.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"48f5a7bba50d0dca353ea142388364028bb8a1e2","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2019-07-24 03:08:05.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"04db44687228494682bce393d0229701bc95c39e","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 03:14:11.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"aebf86052a6f02ff108c4d4028d559301b7670d6","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-24 03:14:11.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/63/658263/6/gate/openstack-tox-docs/d9cdfdf/html/ : SUCCESS in 2m 46s\n- openstack-tox-pep8 http://logs.openstack.org/63/658263/6/gate/openstack-tox-pep8/628912f/ : SUCCESS in 2m 25s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"ce71078ea8af35c9955d84130b007906bf7858d5","revisions":{"a417f0b56f930af77d96dccf77c85cae542addc0":{"kind":"REWORK","_number":1,"created":"2019-05-10 04:50:51.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/63/658263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/63/658263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/1"}}},"commit":{"parents":[{"commit":"b4253edaae5e183d646efae42416ec5bc378e10b","subject":"OpenDev Migration Patch","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b4253edaae5e183d646efae42416ec5bc378e10b"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-10 06:49:52.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-10 06:50:32.000000000","tz":-420},"subject":"Specification for Cyborg APIs Version 2.","message":"Specification for Cyborg APIs Version 2.\n\nChange-Id: I735b296aab1037efdd16eb99d5b5e5da56f95746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a417f0b56f930af77d96dccf77c85cae542addc0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a417f0b56f930af77d96dccf77c85cae542addc0"}]},"branch":"refs/heads/master"},"ff86513612c1e2db21ad79e6afeb4c543d18a708":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2019-06-28 01:29:05.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/63/658263/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/63/658263/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/2"}}},"commit":{"parents":[{"commit":"46b3ef1b49fab156d684eb17fc57534647884e01","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/46b3ef1b49fab156d684eb17fc57534647884e01"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-10 06:49:52.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-06-28 03:24:33.000000000","tz":-420},"subject":"Specification for Cyborg APIs Version 2.","message":"Specification for Cyborg APIs Version 2.\n\nChange-Id: I735b296aab1037efdd16eb99d5b5e5da56f95746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ff86513612c1e2db21ad79e6afeb4c543d18a708"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ff86513612c1e2db21ad79e6afeb4c543d18a708"}]},"branch":"refs/heads/master"},"98e7bbbe29d8c3d226e1e353c538821aac66532b":{"kind":"REWORK","_number":3,"created":"2019-06-28 01:32:50.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/63/658263/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/63/658263/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/3"}}},"commit":{"parents":[{"commit":"46b3ef1b49fab156d684eb17fc57534647884e01","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/46b3ef1b49fab156d684eb17fc57534647884e01"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-10 06:49:52.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-06-28 03:31:20.000000000","tz":-420},"subject":"Specification for Cyborg APIs Version 2.","message":"Specification for Cyborg APIs Version 2.\n\nChange-Id: I735b296aab1037efdd16eb99d5b5e5da56f95746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/98e7bbbe29d8c3d226e1e353c538821aac66532b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/98e7bbbe29d8c3d226e1e353c538821aac66532b"}]},"branch":"refs/heads/master"},"422b81c02d533d88d4638dcc5dd535a826ac0527":{"kind":"REWORK","_number":4,"created":"2019-07-03 21:31:43.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/63/658263/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/63/658263/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/4"}}},"commit":{"parents":[{"commit":"46b3ef1b49fab156d684eb17fc57534647884e01","subject":"Switch to the new canonical constraints URL on master","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/46b3ef1b49fab156d684eb17fc57534647884e01"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-10 06:49:52.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-03 23:30:09.000000000","tz":-420},"subject":"Specification for Cyborg APIs Version 2.","message":"Specification for Cyborg APIs Version 2.\n\nChange-Id: I735b296aab1037efdd16eb99d5b5e5da56f95746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/422b81c02d533d88d4638dcc5dd535a826ac0527"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/422b81c02d533d88d4638dcc5dd535a826ac0527"}]},"branch":"refs/heads/master"},"53e75c038a1d28978158a36d4484f7db99e4c2ad":{"kind":"REWORK","_number":5,"created":"2019-07-12 08:41:31.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/63/658263/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/63/658263/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/5"}}},"commit":{"parents":[{"commit":"2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5","subject":"Merge \"Specification for device profiles.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-10 06:49:52.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 08:39:39.000000000","tz":-420},"subject":"Specification for Cyborg APIs Version 2.","message":"Specification for Cyborg APIs Version 2.\n\nChange-Id: I735b296aab1037efdd16eb99d5b5e5da56f95746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/53e75c038a1d28978158a36d4484f7db99e4c2ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/53e75c038a1d28978158a36d4484f7db99e4c2ad"}]},"branch":"refs/heads/master"},"ce71078ea8af35c9955d84130b007906bf7858d5":{"kind":"REWORK","_number":6,"created":"2019-07-12 08:44:02.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/63/658263/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/63/658263/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/63/658263/6"}}},"commit":{"parents":[{"commit":"2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5","subject":"Merge \"Specification for device profiles.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-10 06:49:52.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-07-12 08:42:56.000000000","tz":-420},"subject":"Specification for Cyborg APIs Version 2.","message":"Specification for Cyborg APIs Version 2.\n\nChange-Id: I735b296aab1037efdd16eb99d5b5e5da56f95746\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ce71078ea8af35c9955d84130b007906bf7858d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ce71078ea8af35c9955d84130b007906bf7858d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
