)]}'
{"id":"openstack%2Fcyborg-specs~658498","triplet_id":"openstack%2Fcyborg-specs~master~I8126104422ff79b444d5ead1a32e1bf49ae882a2","project":"openstack/cyborg-specs","branch":"master","hashtags":[],"change_id":"I8126104422ff79b444d5ead1a32e1bf49ae882a2","subject":"Fix the release name in index.rst.","status":"ABANDONED","created":"2019-05-10 22:40:44.000000000","updated":"2019-06-05 23:48:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b9b458c843922fc2d7a147a912b9c1889f5c3f60","_number":658498,"virtual_id_number":658498,"owner":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":1,"date":"2019-05-10 22:49:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"recommended":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"all":[{"value":2,"date":"2019-05-16 01:18:41.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"value":1,"date":"2019-05-14 12:08:27.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"all":[{"value":1,"date":"2019-05-16 01:18:41.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"date":"2019-05-10 23:29:24.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"value":0,"date":"2019-05-10 23:29:14.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"date":"2019-05-10 23:29:40.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-05-10 22:49:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2019-05-10 23:29:14.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2019-05-10 23:29:24.000000000","updated_by":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"reviewer":{"_account_id":22899,"name":"coco-Gao","email":"419546439@qq.com","username":"Coco"},"state":"REVIEWER"},{"updated":"2019-05-10 23:29:40.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2019-05-14 12:08:27.000000000","updated_by":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"reviewer":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"state":"REVIEWER"},{"updated":"2019-05-16 01:18:41.000000000","updated_by":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"reviewer":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"state":"REVIEWER"}],"messages":[{"id":"aabd474197da0b70b9467628f078f5d77ccef3a7","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-05-10 22:40:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a78f2f166f9818798b6e73f547d35205aeec0ee2","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-05-10 22:49:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/98/658498/1/check/openstack-tox-docs/0768271/html/ : SUCCESS in 3m 58s\n- openstack-tox-pep8 http://logs.openstack.org/98/658498/1/check/openstack-tox-pep8/3900d24/ : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7d29c3a6aeddcea69aa510939f959e6d0077e67c","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-05-14 12:08:27.000000000","message":"Patch Set 1: Code-Review+1\n\nI also fix it in https://review.opendev.org/#/c/651061/","accounts_in_message":[],"_revision_number":1},{"id":"c5d4f44d4a1b9b8649faaa9ca3d881bcba93a07d","author":{"_account_id":27458,"name":"Li Liu","email":"liliueecg@gmail.com","username":"liliu"},"date":"2019-05-16 01:18:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":1},{"id":"fe329f071f8cc4e8e8771712269f92bb3cb65f36","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-06-05 03:55:02.000000000","message":"Patch Set 1:\n\nIt\u0027s already fixed by\nhttps://review.opendev.org/#/c/651061/9/doc/source/index.rst@12","accounts_in_message":[],"_revision_number":1},{"id":"b9b458c843922fc2d7a147a912b9c1889f5c3f60","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-06-05 23:48:39.000000000","message":"Abandoned\n\nAlready fixed by\nhttps://review.opendev.org/#/c/651061/9/doc/source/index.rst@12 . Thanks, Yikun.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b705018fedbcb4a9680dcd58813806ba089652bd","revisions":{"b705018fedbcb4a9680dcd58813806ba089652bd":{"kind":"REWORK","_number":1,"created":"2019-05-10 22:40:44.000000000","uploader":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"ref":"refs/changes/98/658498/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/98/658498/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/658498/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/658498/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/658498/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/98/658498/1"}}},"commit":{"parents":[{"commit":"a417f0b56f930af77d96dccf77c85cae542addc0","subject":"Specification for Cyborg APIs Version 2.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a417f0b56f930af77d96dccf77c85cae542addc0"}]}],"author":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-11 00:39:23.000000000","tz":-420},"committer":{"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","date":"2019-05-11 00:39:23.000000000","tz":-420},"subject":"Fix the release name in index.rst.","message":"Fix the release name in index.rst.\n\nChange-Id: I8126104422ff79b444d5ead1a32e1bf49ae882a2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b705018fedbcb4a9680dcd58813806ba089652bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b705018fedbcb4a9680dcd58813806ba089652bd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
