)]}'
{"id":"openstack%2Fcyborg-specs~669974","triplet_id":"openstack%2Fcyborg-specs~master~Iea6a68c433a1e28db46d1e6635dc2f3e777294a3","project":"openstack/cyborg-specs","branch":"master","hashtags":[],"change_id":"Iea6a68c433a1e28db46d1e6635dc2f3e777294a3","subject":"Move the spec into right location","status":"MERGED","created":"2019-07-10 03:10:45.000000000","updated":"2019-07-26 18:10:34.000000000","submitted":"2019-07-26 18:10:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"669974-1564164634940-cd2898e8","meta_rev_id":"25b6165015c435e2b5ef2b1dace4b04e235e6df3","_number":669974,"virtual_id_number":669974,"owner":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":2,"date":"2019-07-26 18:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":2,"date":"2019-07-24 02:55:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"value":1,"date":"2019-07-26 18:02:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-07-26 18:02:00.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-07-26 18:10:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"470166f7a0e4b3e36b33a9fa8700dd96ec4f3552","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-10 03:10:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fd4171f437b67f5be1f61856af85f14dd0d5297","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-10 03:13:38.000000000","message":"Patch Set 1: Code-Review-1 Workflow-1\n\nshould also fix the format","accounts_in_message":[],"_revision_number":1},{"id":"def4940386233a708327faa6b9d7c16ce5cabd28","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 03:52:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/74/669974/1/check/openstack-tox-docs/488a6b9/html/ : SUCCESS in 3m 14s\n- openstack-tox-pep8 http://logs.openstack.org/74/669974/1/check/openstack-tox-pep8/5e538f5/ : SUCCESS in 3m 06s","accounts_in_message":[],"_revision_number":1},{"id":"83b30da3b1d3891cc63363f5b51e66f361d27412","author":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"date":"2019-07-10 06:20:11.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e67511b2cc6fd5d751cacd9a5eac39efc0e0479b","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-10 06:31:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/74/669974/2/check/openstack-tox-docs/4a9525e/html/ : SUCCESS in 3m 03s\n- openstack-tox-pep8 http://logs.openstack.org/74/669974/2/check/openstack-tox-pep8/0492fe2/ : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":2},{"id":"693b113ee055a8450cfd32b8a2c683ee99f35ef3","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-24 02:55:03.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d2554e2b4fdb1bb0b0f5d63550878e05f338272a","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-07-26 18:02:00.000000000","message":"Patch Set 2: Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"0f76e7d8d04658ec8153354e2e152ff5f87b0cae","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:02:12.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"84481b1208fc930944b97f9787aaa7ba39c7d4f8","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:10:34.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"25b6165015c435e2b5ef2b1dace4b04e235e6df3","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-07-26 18:10:34.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs http://logs.openstack.org/74/669974/2/gate/openstack-tox-docs/3a2ac1c/html/ : SUCCESS in 3m 23s\n- openstack-tox-pep8 http://logs.openstack.org/74/669974/2/gate/openstack-tox-pep8/86f6e9c/ : SUCCESS in 3m 52s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a1d7c2a3848ea01073db9e5ce3cdaa056a4a0e49","revisions":{"2db08d481a780e8a5b346a15117747dc6bbe50fb":{"kind":"REWORK","_number":1,"created":"2019-07-10 03:10:45.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/74/669974/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/74/669974/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/1"}}},"commit":{"parents":[{"commit":"2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5","subject":"Merge \"Specification for device profiles.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-10 03:10:19.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-10 03:10:19.000000000","tz":480},"subject":"Move the spec into right location","message":"Move the spec into right location\n\nChange-Id: Iea6a68c433a1e28db46d1e6635dc2f3e777294a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2db08d481a780e8a5b346a15117747dc6bbe50fb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2db08d481a780e8a5b346a15117747dc6bbe50fb"}]},"branch":"refs/heads/master"},"a1d7c2a3848ea01073db9e5ce3cdaa056a4a0e49":{"kind":"REWORK","_number":2,"created":"2019-07-10 06:20:11.000000000","uploader":{"_account_id":20722,"name":"Yikun Jiang","email":"yikunkero@gmail.com","username":"yikunkero"},"ref":"refs/changes/74/669974/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/74/669974/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/74/669974/2"}}},"commit":{"parents":[{"commit":"2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5","subject":"Merge \"Specification for device profiles.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2dd2aca05fbf5f1d90d8e2e8908bc912dfaf6ce5"}]}],"author":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-10 03:10:19.000000000","tz":480},"committer":{"name":"Yikun Jiang","email":"yikunkero@gmail.com","date":"2019-07-10 06:19:48.000000000","tz":480},"subject":"Move the spec into right location","message":"Move the spec into right location\n\nChange-Id: Iea6a68c433a1e28db46d1e6635dc2f3e777294a3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a1d7c2a3848ea01073db9e5ce3cdaa056a4a0e49"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a1d7c2a3848ea01073db9e5ce3cdaa056a4a0e49"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
