)]}'
{"id":"openstack%2Fcyborg-specs~695840","triplet_id":"openstack%2Fcyborg-specs~master~If64971a3a0cfd97d2597654a3c8896af1af3b33a","project":"openstack/cyborg-specs","branch":"master","hashtags":[],"change_id":"If64971a3a0cfd97d2597654a3c8896af1af3b33a","subject":"Re-propose specification for Cyborg APIs Version 2","status":"MERGED","created":"2019-11-25 05:58:34.000000000","updated":"2019-12-15 10:09:11.000000000","submitted":"2019-12-15 10:07:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":15,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"695840-1576404478399-91269b4c","meta_rev_id":"ed2ed36d435932066345d5e81265b3fc1238364c","_number":695840,"virtual_id_number":695840,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"date":"2019-12-15 09:56:11.000000000","_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2019-12-15 10:07:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":2,"date":"2019-12-15 09:02:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":1,"date":"2019-12-10 11:49:28.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2019-12-15 06:27:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"all":[{"value":0,"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":1,"date":"2019-12-15 09:02:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-11-26 03:32:57.000000000","updated_by":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"reviewer":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"state":"REVIEWER"},{"updated":"2019-12-03 09:31:33.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2019-12-10 11:49:28.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2019-12-15 06:27:31.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2019-12-15 09:02:34.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2019-12-15 10:07:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"78d2c607ce00751a8bcf536934f6d14978db4c5e","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-11-25 05:58:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b9cb064cd73ba2022bd953f66d673c4d14728756","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 06:06:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43e5b7990c744221bd4f5e133efb09c6 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8161e20121554a53ae0f5db317830703 : SUCCESS in 2m 59s","accounts_in_message":[],"_revision_number":1},{"id":"3e5ea2297b282180c751e35d5aad7a0d68fb3940","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-11-25 06:45:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"ede9369e8d48c924fed63f4e6cf5b9d11d65d3b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-25 06:58:49.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b8caed12b7410187ddba529f15cb18 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53701e375de4545b30e33d613523a4c : SUCCESS in 4m 05s","accounts_in_message":[],"_revision_number":2},{"id":"7f7b29a7d714b4b9320594178fdfca4970ecba85","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-11-25 23:45:32.000000000","message":"Patch Set 2: Code-Review-1\n\nThanks but the right process is this:\n* The change to doc/source/index.rst for Ussuri should be a separate patch. \n* The old spec from Train should be re-proposed exactly as is, in a separate patch. Like what I did in Nova [1].\n* Any changes from the old spec will be a 3rd patch.\n\nThis will allow the changes to be reviewed easily. These are the common conventions in other projects too.\n\n[1] https://review.opendev.org/#/c/682637/","accounts_in_message":[],"_revision_number":2},{"id":"ae300c3104ed9be465c5a8c3f05282c406b49b85","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-11-26 02:48:37.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0e1000048e9693c74e95d89d44c3d06d0444e55b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 02:56:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc59cda55c66407d869906ea69f2b95d : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2df380d0cee4244afbd825fd23ecd04 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"36cc4c3015ff929abef9a97d51292f9bc4b227bc","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-11-26 04:35:21.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)\n\nNeeds +2 from non-Intel core, then I\u0027ll give +2.","accounts_in_message":[],"_revision_number":3},{"id":"908a5bd4da6ce3dc96038332cbe6f69a82b0d323","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-11-26 05:14:46.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"a6a18627576b0eba0171574a8bc5fb3e4b20539c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-11-26 05:18:16.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e23ba6b9a224dd2aedc7e885d2d16b8 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2eb56dd3582c4174ac20e9b2681165d8 : SUCCESS in 2m 33s","accounts_in_message":[],"_revision_number":4},{"id":"3e0d6926e662e2be6b23f9788acb2e18312feabd","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-12-03 09:31:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d98d44059fa474d696a3a874e473568d24221aff","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-04 03:34:58.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nthanks xinran. overall looks good to me. Soft -1 for a little format error.","accounts_in_message":[],"_revision_number":4},{"id":"eb4239d54027796359da4cd4cd9e8a29f770e5ba","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-05 07:21:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"2b6012f00744d99e2b0b4f6ac25354ee089ba4b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-05 07:26:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fe608b33f254b1a81e33c5fd87df862 : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1468c6f7b3b439bbf974d0a124d7983 : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":5},{"id":"bd8840f8ca22d0cfbabab555da3180fb59f19e2e","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-06 02:12:10.000000000","message":"Patch Set 4:\n\n(5 comments)\n\nThanks for your comments Chenke.","accounts_in_message":[],"_revision_number":4},{"id":"0947f23807fff86ae1040ae9db3f10341386c869","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-12-06 06:42:39.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nhi Xinran, pls see my comments inline.\nsoft -1.","accounts_in_message":[],"_revision_number":4},{"id":"b9ca33c115875b9c65abaec62dafb135275cb043","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-12-06 07:00:12.000000000","message":"Patch Set 5: Code-Review-1\n\nops, sorry Xinran. \nthe following comments should be in patch5.\nsoft -1.\n \u003e (2 comments)\nspecs/ussuri/approved/cyborg-api.rst\nLine 79:\nBy saying \"the deployable object is exposed as part of\nthe device object\" here, shall I suppose that the GET /v2/devices request also returns deployable info? and another question: Is it necessary for cyborg to expose the deployable objects to users? For example, if cyborg manage an FPGA with two regions, in the object level, that\u0027s one device including two deployables, right? How does the user know that\u0027s a device two regions?\n\nLine 107:\nHi Xinran, Just to check if this is what you mentioned to be done here?\nhttps://review.opendev.org/#/c/677115/16/cyborg/db/sqlalchemy/models.py","accounts_in_message":[],"_revision_number":5},{"id":"ccc0da261cbe4feb237140d25aea10fc9261a63e","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-06 07:14:54.000000000","message":"Patch Set 4:\n\n(2 comments)\n\n\u003e ops, sorry Xinran.\n \u003e the following comments should be in patch5.\n \u003e soft -1.\n \u003e \u003e (2 comments)\n \u003e specs/ussuri/approved/cyborg-api.rst\n \u003e Line 79:\n \u003e By saying \"the deployable object is exposed as part of\n \u003e the device object\" here, shall I suppose that the GET /v2/devices\n \u003e request also returns deployable info? and another question: Is it\n \u003e necessary for cyborg to expose the deployable objects to users? For\n \u003e example, if cyborg manage an FPGA with two regions, in the object\n \u003e level, that\u0027s one device including two deployables, right? How does\n \u003e the user know that\u0027s a device two regions?\n \u003e \n \u003e Line 107:\n \u003e Hi Xinran, Just to check if this is what you mentioned to be done\n \u003e here?\n \u003e https://review.opendev.org/#/c/677115/16/cyborg/db/sqlalchemy/models.py\n\nHi Yumeng,  thanks for your comments. \nI\u0027d like to mention that this spec is merged in train, I just re-propose it in ussuri without important changes. After this merged, I will work on https://review.opendev.org/#/c/696012/ to update it.","accounts_in_message":[],"_revision_number":4},{"id":"fdcecacaa211776d1e6262b0bd4ecf2a78141dde","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2019-12-10 11:49:28.000000000","message":"Patch Set 5: Code-Review+1\n\n\u003e (2 comments)\n \u003e \n \u003e \u003e ops, sorry Xinran.\n \u003e \u003e the following comments should be in patch5.\n \u003e \u003e soft -1.\n \u003e \u003e \u003e (2 comments)\n \u003e \u003e specs/ussuri/approved/cyborg-api.rst\n \u003e \u003e Line 79:\n \u003e \u003e By saying \"the deployable object is exposed as part of\n \u003e \u003e the device object\" here, shall I suppose that the GET /v2/devices\n \u003e \u003e request also returns deployable info? and another question: Is it\n \u003e \u003e necessary for cyborg to expose the deployable objects to users?\n \u003e For\n \u003e \u003e example, if cyborg manage an FPGA with two regions, in the object\n \u003e \u003e level, that\u0027s one device including two deployables, right? How\n \u003e does\n \u003e \u003e the user know that\u0027s a device two regions?\n \u003e \u003e\n \u003e \u003e Line 107:\n \u003e \u003e Hi Xinran, Just to check if this is what you mentioned to be done\n \u003e \u003e here?\n \u003e \u003e https://review.opendev.org/#/c/677115/16/cyborg/db/sqlalchemy/models.py\n \u003e \n \u003e Hi Yumeng,  thanks for your comments.\n \u003e I\u0027d like to mention that this spec is merged in train, I just\n \u003e re-propose it in ussuri without important changes. After this\n \u003e merged, I will work on https://review.opendev.org/#/c/696012/ to\n \u003e update it.\n\nok. remove soft -1 and let\u0027s see if others have comments.","accounts_in_message":[],"_revision_number":5},{"id":"8b15a15dc462854ccfa8412b71f8b51daa4533bc","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-15 06:27:31.000000000","message":"Patch Set 5: Code-Review+2\n\nThanks xinran. LGTM.","accounts_in_message":[],"_revision_number":5},{"id":"eadda70a6f031d1e24b774475d6b010937d7eb7b","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-12-15 09:02:34.000000000","message":"Patch Set 5: Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"8c00a84155510d3d2f5f94019449645ba460ff6e","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2019-12-15 09:02:41.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c0998ef1dd388e22150d9f02c515a129e579c14a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 09:02:45.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"9c764d93d36419f8c069ce66950b1720225931a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 09:07:29.000000000","message":"Patch Set 5: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d2e04353f0c44395b5cbaa7f31e59ce0 : FAILURE in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fb399f5770c43b28d4c80a4f778d35c : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":5},{"id":"15ee9b3d4d3de9bc61a43a385658ed0e345324ca","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-15 09:25:52.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"53e316477b60d2074c450fd325a8804d2712d2bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 09:41:39.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74cfbe9a7d874f1fbc9859824e92d04b : FAILURE in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b54ba5c8ffb425696080c9859da66a2 : SUCCESS in 3m 12s","accounts_in_message":[],"_revision_number":5},{"id":"428ef0b109f9f78f0591444bf6cee7e0f7aa19b1","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-15 09:56:11.000000000","message":"Patch Set 5:\n\nrecheck","accounts_in_message":[],"_revision_number":5},{"id":"814ca61224e536ff322ccfd79c7f9b1f5c327985","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 09:59:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cbd2aaccbb704ba9bc924cf621449240 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/85918ee61184469c92a9e7a817c73a6b : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"2160019995992dfcdaefc6ec7b053116c6f94317","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 09:59:55.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"c0021eb544ae55d29cbb24b3424158d2a3264fd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 10:07:58.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/304b5a9bf7ea48459ed58cc9ddec1664 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/76bac06714754c3baddd068344f18953 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":5},{"id":"2473412c998ecfb41b1e426dacc67858f1240d04","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 10:07:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":5},{"id":"ed2ed36d435932066345d5e81265b3fc1238364c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-15 10:09:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/9ee69e74f4a4484b854cd68c37f43dad : SUCCESS in 57s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"3274a22dec864b280c67cf8147d187827a1daba3","revisions":{"c5b94f030b9f028d552440e647a9171917e4cb44":{"kind":"REWORK","_number":1,"created":"2019-11-25 05:58:34.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/40/695840/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/40/695840/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/1"}}},"commit":{"parents":[{"commit":"73ea25f3c54552d442d0218f4cc11929d1425530","subject":"Merge \"Move the spec into right location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/73ea25f3c54552d442d0218f4cc11929d1425530"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-25 05:45:45.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-25 05:54:27.000000000","tz":480},"subject":"Specification for Device V2 API.","message":"Specification for Device V2 API.\n\nChange-Id: If64971a3a0cfd97d2597654a3c8896af1af3b33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/c5b94f030b9f028d552440e647a9171917e4cb44"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/c5b94f030b9f028d552440e647a9171917e4cb44"}]},"branch":"refs/heads/master"},"45764c1d1f51075c77bcd4dacac451a0eab5ecf3":{"kind":"REWORK","_number":2,"created":"2019-11-25 06:45:27.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/40/695840/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/40/695840/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/2"}}},"commit":{"parents":[{"commit":"73ea25f3c54552d442d0218f4cc11929d1425530","subject":"Merge \"Move the spec into right location\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/73ea25f3c54552d442d0218f4cc11929d1425530"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-25 05:45:45.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-25 06:41:09.000000000","tz":480},"subject":"Specification for Device V2 API.","message":"Specification for Device V2 API.\n\nChange-Id: If64971a3a0cfd97d2597654a3c8896af1af3b33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/45764c1d1f51075c77bcd4dacac451a0eab5ecf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/45764c1d1f51075c77bcd4dacac451a0eab5ecf3"}]},"branch":"refs/heads/master"},"87055d961ca16130cb286dbcda3b18d5c504775f":{"kind":"REWORK","_number":3,"created":"2019-11-26 02:48:37.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/40/695840/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/40/695840/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/3"}}},"commit":{"parents":[{"commit":"beb8723fffb01f64f18d5ea335540da4d778e1e0","subject":"Add Ussuri spec path in doc/index.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/beb8723fffb01f64f18d5ea335540da4d778e1e0"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-25 05:45:45.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-26 02:43:37.000000000","tz":480},"subject":"Re-propose specification for Cyborg APIs Version 2","message":"Re-propose specification for Cyborg APIs Version 2\n\nChange-Id: If64971a3a0cfd97d2597654a3c8896af1af3b33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/87055d961ca16130cb286dbcda3b18d5c504775f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/87055d961ca16130cb286dbcda3b18d5c504775f"}]},"branch":"refs/heads/master"},"52149e0c5a02bf1c1c76353067c07b1781b2c6d5":{"kind":"REWORK","_number":4,"created":"2019-11-26 05:14:46.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/40/695840/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/40/695840/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/4"}}},"commit":{"parents":[{"commit":"beb8723fffb01f64f18d5ea335540da4d778e1e0","subject":"Add Ussuri spec path in doc/index.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/beb8723fffb01f64f18d5ea335540da4d778e1e0"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-25 05:45:45.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2019-11-26 05:14:40.000000000","tz":0},"subject":"Re-propose specification for Cyborg APIs Version 2","message":"Re-propose specification for Cyborg APIs Version 2\n\nChange-Id: If64971a3a0cfd97d2597654a3c8896af1af3b33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/52149e0c5a02bf1c1c76353067c07b1781b2c6d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/52149e0c5a02bf1c1c76353067c07b1781b2c6d5"}]},"branch":"refs/heads/master"},"3274a22dec864b280c67cf8147d187827a1daba3":{"kind":"REWORK","_number":5,"created":"2019-12-05 07:21:26.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/40/695840/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/40/695840/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/40/695840/5"}}},"commit":{"parents":[{"commit":"47193f74d50ed6a33b01ba2e8053ea9b7c304c38","subject":"Add Ussuri spec path in doc/index.rst","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/47193f74d50ed6a33b01ba2e8053ea9b7c304c38"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-11-25 05:45:45.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-05 07:17:29.000000000","tz":480},"subject":"Re-propose specification for Cyborg APIs Version 2","message":"Re-propose specification for Cyborg APIs Version 2\n\nChange-Id: If64971a3a0cfd97d2597654a3c8896af1af3b33a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3274a22dec864b280c67cf8147d187827a1daba3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3274a22dec864b280c67cf8147d187827a1daba3"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
