)]}'
{"id":"openstack%2Fcyborg-specs~699149","triplet_id":"openstack%2Fcyborg-specs~master~I9b5f02daaf91ecd4416f032834c4fafdcb003b36","project":"openstack/cyborg-specs","branch":"master","topic":"api-microversioning","hashtags":[],"change_id":"I9b5f02daaf91ecd4416f032834c4fafdcb003b36","subject":"Specification of Cyborg API microversion","status":"MERGED","created":"2019-12-16 03:34:21.000000000","updated":"2020-02-03 09:45:30.000000000","submitted":"2020-02-03 09:43:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":81,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"699149-1580723001984-159bbb0d","meta_rev_id":"2fb495a300e6892b049b6d10e055c42f51167396","_number":699149,"virtual_id_number":699149,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-02-03 09:43:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":30751,"name":"Shogo Saito","email":"shogosaito438@gmail.com","username":"s_shogo"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2020-02-03 09:34:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":1,"date":"2020-02-03 07:11:03.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-02-03 07:22:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":30751,"name":"Shogo Saito","email":"shogosaito438@gmail.com","username":"s_shogo"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2020-02-03 09:34:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":30751,"name":"Shogo Saito","email":"shogosaito438@gmail.com","username":"s_shogo"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":30751,"name":"Shogo Saito","email":"shogosaito438@gmail.com","username":"s_shogo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2019-12-16 03:41:10.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2019-12-16 03:41:17.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2019-12-20 10:35:05.000000000","updated_by":{"_account_id":30751,"name":"Shogo Saito","email":"shogosaito438@gmail.com","username":"s_shogo"},"reviewer":{"_account_id":30751,"name":"Shogo Saito","email":"shogosaito438@gmail.com","username":"s_shogo"},"state":"REVIEWER"},{"updated":"2020-01-21 09:53:25.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-02-03 07:11:03.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-02-03 07:22:03.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-02-03 09:34:40.000000000","updated_by":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"reviewer":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"state":"REVIEWER"},{"updated":"2020-02-03 09:43:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a9d21cc78c42b322841902fe1580c77a001e72ef","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-16 03:34:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3c6811de8464550dca688ac5399cfe1dcc666f9b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-16 03:40:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5944e03f41404cd8a89be80cfa6426b8 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6056dbfa0d824e1a89fc5327f6d4a66b : SUCCESS in 4m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0149ab97902cf69cbecd73edb9ad12bf0e774d61","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-19 14:41:11.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nThanks xinran. After read the http://specs.openstack.org/openstack/api-wg/guidelines/microversion_specification.html and https://specs.openstack.org/openstack/cinder-specs/specs/mitaka/api-microversions.html. This patch LGTM. soft -1, Pls see my comment inline.","accounts_in_message":[],"_revision_number":1},{"id":"aaa82d958a3e87234b64daaa4a4e70a82243cff1","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-12-20 03:54:45.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nLooks good overall.","accounts_in_message":[],"_revision_number":1},{"id":"e93b9f093c4c81ae4b8cc5b39bb68c2ce711b163","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-20 05:36:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff015e8d64915bb09663213ffd7c6fac82a8c57d","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2019-12-20 10:36:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b52646d68a274dda3f91e5a5752d09b8dd244770","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-24 06:45:40.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"f18cdd2aaf6813b38a0c202da80ff1e7ce22b278","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-25 02:15:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a42b31797f0b03725add041f3bdd9c94cde91cbd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-25 02:19:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91ef600354f445dd9d5784bd664a15ba : SUCCESS in 3m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1898ff0e8b149229538c1d3e5fb25d5 : SUCCESS in 3m 01s","accounts_in_message":[],"_revision_number":2},{"id":"e451a2011875487c1c9f513d775027f137978424","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-30 09:18:16.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)\n\nOverall LGTM.","accounts_in_message":[],"_revision_number":2},{"id":"9dab060b2e094de85519af20238d78e25eb6308f","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-30 09:18:46.000000000","message":"Patch Set 2:\n\nvscode/setting.json needs to be removed.","accounts_in_message":[],"_revision_number":2},{"id":"45f6c7743a8ec2c1fdded77d0d32a405a43038b5","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2019-12-31 02:14:55.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3baf50203873918676e41305f6b761689498261","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2019-12-31 02:17:03.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"ea22e0886c1ebd4dfc3c9968e30f98304fb00e22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2019-12-31 02:18:31.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8b8b1ce0c5e42908b0a8e02593bab1c : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/39c8a73c10ad48dcb21ffd1ac678547e : SUCCESS in 2m 55s","accounts_in_message":[],"_revision_number":3},{"id":"28ffde980db88cf6af5e213906622c55344b3195","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-01-02 13:58:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)\n\nIn general this is conceptually fine and gets the main principles of microversions. However, the header is incorrect and should be fixed.\n\nI\u0027ve also added some notes about code to use.","accounts_in_message":[],"_revision_number":3},{"id":"b64da88ec853d69071c4825ed9c1bf039ef038c9","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-06 03:13:44.000000000","message":"Patch Set 3:\n\n(5 comments)\n\nHi Chris, thanks for your comments, please see my reply inline :)","accounts_in_message":[],"_revision_number":3},{"id":"6eb46c7e94c4937bb507c73d666b01956d9d4284","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-01-06 23:11:37.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"506a01aa2bf29d8336801b8b1766e2e379cfbf29","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-07 06:33:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"94f6f4c960f4f3f3fc17d341a4babe3908d91ce2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-07 07:54:18.000000000","message":"Patch Set 3: Code-Review-1\n\n(12 comments)\n\nSoft -1, this change looks good to me, some suggestions inline.","accounts_in_message":[],"_revision_number":3},{"id":"f3eb62ca1920eb74e0dbdf6a015f386b7ceafa36","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-07 08:04:57.000000000","message":"Removed Code-Review+2 by chenker \u003cchen.ke14@zte.com.cn\u003e\n","accounts_in_message":[],"_revision_number":3},{"id":"f4342486b5779f5881a239311e822078225539ec","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-07 08:11:21.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"044326ff3ab1f522932c4215cbc797d9c5b018e1","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-13 08:03:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"5b94caaf9196ef941cffb6e7a85e38e116b0df98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-13 08:08:01.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ce5599687ca49f8bb4fd6108cea35a9 : SUCCESS in 3m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/153ea67410084b05b0caa425bfb12307 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":4},{"id":"09c25fd589646add115f029a325d6c6f3c770214","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-14 03:38:48.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"f1b6990c6fd0db3a1b46fe47b19653f61e625c43","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-14 10:29:40.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)\n\nThanks for your updated, LGTM now.","accounts_in_message":[],"_revision_number":4},{"id":"07fe425b5c2da192311a229a2f51746f81bf6c10","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-14 11:27:38.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"1001c3b63d7f46c3a167e910d0498d14fab3f68b","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-19 02:30:49.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n \u003e \n \u003e Thanks for your updated, LGTM now.\n\nIt is for developer who use VSCode as the editor, we should ignore this directory when submit a patch.","accounts_in_message":[],"_revision_number":4},{"id":"d2cc877782fe669b30f60ccd932e7b40ebf4375d","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-19 02:31:23.000000000","message":"Patch Set 4:\n\n\u003e (1 comment)\n\nYes, agree with you. will update commit msg.","accounts_in_message":[],"_revision_number":4},{"id":"ac600ed24a67a6ae0a477e4910242c1c472baa0d","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-19 03:05:35.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"eeed1c2eb08d6d9861633e80754e9a9e7e83a79f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 03:13:55.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1862fad08c8846f68d7c1789d7f37101 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d361a1adfc7540b5a5ecd21b0a81690d : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":5},{"id":"42879e9e6f3797813ee242868a48be6feae97a3a","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 03:25:18.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"0cb80280360fd871b8e586f2dd4d0c14eef8ba3c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-19 03:29:51.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)\n\nThanks, LGTM","accounts_in_message":[],"_revision_number":5},{"id":"de06ebf76fd167c72eae3eacc8b8c2ddea32b6a0","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-19 03:35:37.000000000","message":"Patch Set 6: Commit message was updated.","accounts_in_message":[],"_revision_number":6},{"id":"b4643da524d99af2e0beab84634e6a2dd17643ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-19 03:39:24.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57cf00aae5654ec3b14a704efa2c163c : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bea60ad26c2f4680b7702d46b5d816a7 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":6},{"id":"b7698467168ab8d82cff57242177db992d04feff","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-01-19 03:41:12.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f29c92d0c1958c20c0081e73463843dfea246e92","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-21 03:01:30.000000000","message":"Patch Set 6: Code-Review-1\n\n(11 comments)\n\nSoft -1, thanks update.\nLGTM in general, but some concern inline. After reply or updated my comments, I will +1 for this.","accounts_in_message":[],"_revision_number":6},{"id":"1648ceb1b82d56c7d984a229f2ca0dbd5dd50af0","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-01-21 09:53:25.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)\n\nThanks Xinran!\nLGTM.","accounts_in_message":[],"_revision_number":6},{"id":"ba86ceb59c841995ce65e652992ac98eb0085a14","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-21 10:49:22.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b0ca20c56d93da04dc18f425a5614fefe7146437","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-21 10:50:27.000000000","message":"Patch Set 7:\n\n(1 comment)\n\n\u003e (1 comment)\n \u003e \n \u003e Thanks Xinran!\n \u003e LGTM.\n\nThanks for your comments, I have update a new version according to other\u0027s comments, please review it ;)","accounts_in_message":[],"_revision_number":7},{"id":"80749273e707b11d4405661479432bcfa46d423b","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-21 10:50:41.000000000","message":"Patch Set 7:\n\n\u003e Patch Set 6: Code-Review+2\n\nThanks for your comments, I have update a new version according to other\u0027s comments, please review it ;)","accounts_in_message":[],"_revision_number":7},{"id":"a420457c6c49d2ec3fee5392e2dc714bd35d2c76","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-21 10:53:30.000000000","message":"Patch Set 6:\n\n(10 comments)\n\nHi Brin, please see my updates, thanks :)","accounts_in_message":[],"_revision_number":6},{"id":"e5864d3870b3633fc8d1275287954658d35c442e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-21 10:55:42.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttp://docs.openstack.org/infra/manual/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e3e2f538aff4a80b3a3d8491598d03a : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8d7132de2074549841376cf8d9d19d3 : FAILURE in 5m 39s","accounts_in_message":[],"_revision_number":7},{"id":"0e1efbf94df1b5d881dc122256b024911033b66e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-21 23:55:34.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"28c265a43322c33a9ca8e69d09ad26bc8bf9e8a8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-22 00:22:10.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"25b55f8222b1b85c27b3001fc00638334f72ebae","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-01-22 01:01:32.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"92973e1bdd435cac77a6cedb2513d84b3446ff3b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 01:14:35.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d41c2a94f6c144e0af4668d60452fa1a : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccb3a39a20a0446784bcf4b4f8d0b498 : SUCCESS in 4m 16s","accounts_in_message":[],"_revision_number":8},{"id":"534389e5bd655db1d4d6a7390753da6424a79549","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-22 01:28:26.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"bf69ea873f822814b314bac4b8e208d0dde6502b","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-22 01:30:02.000000000","message":"Patch Set 9: Code-Review+1\n\nLGTM, I just fine-tuned the format.","accounts_in_message":[],"_revision_number":9},{"id":"f95e6a71b5142296e8677e5b003972e48ae2e1af","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-01-22 01:35:35.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3e23b163efa430e8ae40e014ebaadb2 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48d4d567369b4012bf81bd0ccfe3a067 : SUCCESS in 4m 39s","accounts_in_message":[],"_revision_number":9},{"id":"8b70afd0c811a3bc0c8157d5c064bce7f8083090","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-01-29 03:46:31.000000000","message":"Patch Set 9: Code-Review-1\n\n(5 comments)\n\nMostly ok with the content, except for reporting v1 API. Chris Dent, hope you can weigh in on whether v1 needs to be reported when we don\u0027t intend to support it in Ussuri (we marked it as deprecated in Train).\n\nSome minor updates to the text.","accounts_in_message":[],"_revision_number":9},{"id":"4d9192298526b600af1dcc14c4b2249459eb5e04","author":{"_account_id":11564,"name":"Chris Dent","email":"cdent@anticdent.org","username":"chdent"},"date":"2020-01-29 10:12:24.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b76240d64607cfc01531aa80b7166db82581d4cd","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-01-30 09:48:45.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"511039425db399f45cfd2e3717a36cbb00c03a4c","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-01-30 19:43:10.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d467420e5d24a57d7ee224b3dba32bd3c49d0f66","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-01 09:27:48.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"d890ee3adb39ff38a61bbbdc3db1b6e0ff9d1b5b","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-02-03 03:20:13.000000000","message":"Patch Set 9:\n\n(2 comments)\n\nThanks all, will update it soon.","accounts_in_message":[],"_revision_number":9},{"id":"971b3eb54d418e90573cf73f79a2eeb4578557c6","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-02-03 06:12:20.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"d6e19a7e34a03620e0d4fe7a0d3fb2e052e612a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 06:23:00.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e9b3ecbf0e4105ac43c4214f366c1f : SUCCESS in 5m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a10494fdb814dc79518000d8663b932 : SUCCESS in 4m 48s","accounts_in_message":[],"_revision_number":10},{"id":"3944a9b3fdafccc28dc50e12563e6e02bfb0d8d8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-02-03 07:11:03.000000000","message":"Patch Set 10: Code-Review+1\n\nLGTM","accounts_in_message":[],"_revision_number":10},{"id":"7111957ef580628a008187e324fbe13dcb946e6d","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-02-03 07:22:03.000000000","message":"Patch Set 10: Code-Review+2\n\nSeems already fixed all advice. LGTM.","accounts_in_message":[],"_revision_number":10},{"id":"25d1914cf81c32458d9f420a15539e4ee93e0f47","author":{"_account_id":21672,"name":"Sundar Nadathur","email":"sundar.nadathur@intel.com","username":"nsundar"},"date":"2020-02-03 09:34:40.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"0dbb9bea6b3922d7697f83f4044820f41affbde1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 09:34:52.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"48e392af7b87a2568a64483864361cf1c06511b1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 09:43:21.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e193a06001e4761bf7eed891ca9d665 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d7ce12bdd244e85b585251d23220a51 : SUCCESS in 6m 23s","accounts_in_message":[],"_revision_number":10},{"id":"f881420ace2e3b771506658d629b2461e27ffae7","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 09:43:21.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"394b83fbb29ce6bf7d0ea4c686412273ee06df58","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-02-03 09:45:30.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/297745b9e1624b538d0233e5ac60e82d : SUCCESS in 1m 34s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"f6d309eed6f0bacce3393553e04265201c2cac55","revisions":{"a6623e31b5e28dbda5828ec2a8539d3fa2b0a4b5":{"kind":"REWORK","_number":1,"created":"2019-12-16 03:34:21.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/1"}}},"commit":{"parents":[{"commit":"3274a22dec864b280c67cf8147d187827a1daba3","subject":"Re-propose specification for Cyborg APIs Version 2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3274a22dec864b280c67cf8147d187827a1daba3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"subject":"Specification of Cyborg API microversion.","message":"Specification of Cyborg API microversion.\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a6623e31b5e28dbda5828ec2a8539d3fa2b0a4b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a6623e31b5e28dbda5828ec2a8539d3fa2b0a4b5"}]},"branch":"refs/heads/master"},"29b1ceeeccfbc4028bfd57bdd1df88f5cb2574b9":{"kind":"REWORK","_number":2,"created":"2019-12-25 02:15:57.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/2"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-25 02:10:58.000000000","tz":480},"subject":"Specification of Cyborg API microversion.","message":"Specification of Cyborg API microversion.\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/29b1ceeeccfbc4028bfd57bdd1df88f5cb2574b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/29b1ceeeccfbc4028bfd57bdd1df88f5cb2574b9"}]},"branch":"refs/heads/master"},"b637b3245ea9c9fb5a63594ec81a91b7104d96f8":{"kind":"REWORK","_number":3,"created":"2019-12-31 02:14:55.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/3"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-31 02:09:44.000000000","tz":480},"subject":"Specification of Cyborg API microversion.","message":"Specification of Cyborg API microversion.\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b637b3245ea9c9fb5a63594ec81a91b7104d96f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b637b3245ea9c9fb5a63594ec81a91b7104d96f8"}]},"branch":"refs/heads/master"},"a33f3b53b59ec832774710fc30a919c1c5f29c01":{"kind":"REWORK","_number":4,"created":"2020-01-13 08:03:12.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/4"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2020-01-13 07:57:52.000000000","tz":480},"subject":"Specification of Cyborg API microversion.","message":"Specification of Cyborg API microversion.\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a33f3b53b59ec832774710fc30a919c1c5f29c01"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a33f3b53b59ec832774710fc30a919c1c5f29c01"}]},"branch":"refs/heads/master"},"9b3d6178b89b939950092bfbebc863b4be62ad2c":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2020-01-19 03:05:35.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/5"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2020-01-19 03:05:22.000000000","tz":0},"subject":"Specification of Cyborg API microversion.","message":"Specification of Cyborg API microversion.\n\nMany changes to the Cyborg REST API require changes to the consumers of\nthe API.\nFor example, If we need to add a required parameter to a method that is\ncalled by Nova, we\u0027d need both the Nova calling code and the\ncyborgclient that Nova uses to change. But newer Cyborg versions with\nthe change must work with older Nova versions, and there is no mechanism\nfor this at the moment. Adding microversions will solve this problem.\n\nAPIImpact\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9b3d6178b89b939950092bfbebc863b4be62ad2c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9b3d6178b89b939950092bfbebc863b4be62ad2c"}]},"branch":"refs/heads/master"},"5e0ccd9c128741803dd785a7b0cf45397d9fb3aa":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2020-01-19 03:35:37.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/6"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2020-01-19 03:35:33.000000000","tz":0},"subject":"Specification of Cyborg API microversion.","message":"Specification of Cyborg API microversion.\n\nMany changes to the Cyborg REST API require changes to the consumers of\nthe API.\nFor example, if we need to add a required parameter to a method that is\ncalled by Nova, we\u0027d need both the Nova calling code and the\ncyborgclient that Nova uses to change. But newer Cyborg versions with\nthe change must work with older Nova versions, and there is no mechanism\nfor this at the moment. Adding microversions will solve this problem.\n\nAPIImpact\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5e0ccd9c128741803dd785a7b0cf45397d9fb3aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5e0ccd9c128741803dd785a7b0cf45397d9fb3aa"}]},"branch":"refs/heads/master"},"29076921da057177f66be4809633c4a919ec2adf":{"kind":"REWORK","_number":7,"created":"2020-01-21 10:49:22.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/7"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2020-01-21 10:43:01.000000000","tz":480},"subject":"Specification of Cyborg API microversion","message":"Specification of Cyborg API microversion\n\nMany changes to the Cyborg REST API require changes to the consumers of\nthe API.\nFor example, if we need to add a required parameter to a method that is\ncalled by Nova, we\u0027d need both the Nova calling code and the\ncyborgclient that Nova uses to change. But newer Cyborg versions with\nthe change must work with older Nova versions, and there is no mechanism\nfor this at the moment. Adding microversions will solve this problem.\n\nAPIImpact\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/29076921da057177f66be4809633c4a919ec2adf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/29076921da057177f66be4809633c4a919ec2adf"}]},"branch":"refs/heads/master"},"50087ea53275b58096b248413c57864765808503":{"kind":"REWORK","_number":8,"created":"2020-01-22 01:01:32.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/8"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2020-01-22 00:55:54.000000000","tz":480},"subject":"Specification of Cyborg API microversion","message":"Specification of Cyborg API microversion\n\nMany changes to the Cyborg REST API require changes to the consumers of\nthe API.\nFor example, if we need to add a required parameter to a method that is\ncalled by Nova, we\u0027d need both the Nova calling code and the\ncyborgclient that Nova uses to change. But newer Cyborg versions with\nthe change must work with older Nova versions, and there is no mechanism\nfor this at the moment. Adding microversions will solve this problem.\n\nAPIImpact\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/50087ea53275b58096b248413c57864765808503"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/50087ea53275b58096b248413c57864765808503"}]},"branch":"refs/heads/master"},"bc7d44918c74b738eee4596eb1fac2e7d67fb477":{"kind":"REWORK","_number":9,"created":"2020-01-22 01:28:26.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/49/699149/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/9"}}},"commit":{"parents":[{"commit":"e31656063ae8d44f8fed8b62c4ccca9b8dc241a3","subject":"Update device section in Cyborg V2 APIs specification.","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e31656063ae8d44f8fed8b62c4ccca9b8dc241a3"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-01-22 01:28:21.000000000","tz":480},"subject":"Specification of Cyborg API microversion","message":"Specification of Cyborg API microversion\n\nMany changes to the Cyborg REST API require changes to the consumers of\nthe API.\nFor example, if we need to add a required parameter to a method that is\ncalled by Nova, we\u0027d need both the Nova calling code and the\ncyborgclient that Nova uses to change. But newer Cyborg versions with\nthe change must work with older Nova versions, and there is no mechanism\nfor this at the moment. Adding microversions will solve this problem.\n\nAPIImpact\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/bc7d44918c74b738eee4596eb1fac2e7d67fb477"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/bc7d44918c74b738eee4596eb1fac2e7d67fb477"}]},"branch":"refs/heads/master"},"f6d309eed6f0bacce3393553e04265201c2cac55":{"kind":"REWORK","_number":10,"created":"2020-02-03 06:12:20.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/49/699149/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/49/699149/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/49/699149/10"}}},"commit":{"parents":[{"commit":"452028b602913a191a819bc2158c0595caf85628","subject":"Move basepython variable to the [testenv]","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/452028b602913a191a819bc2158c0595caf85628"}]}],"author":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2019-12-16 03:26:06.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"wxr930217@gmail.com","date":"2020-02-03 06:02:39.000000000","tz":480},"subject":"Specification of Cyborg API microversion","message":"Specification of Cyborg API microversion\n\nMany changes to the Cyborg REST API require changes to the consumers of\nthe API.\nFor example, if we need to add a required parameter to a method that is\ncalled by Nova, we\u0027d need both the Nova calling code and the\ncyborgclient that Nova uses to change. But newer Cyborg versions with\nthe change must work with older Nova versions, and there is no mechanism\nfor this at the moment. Adding microversions will solve this problem.\n\nAPIImpact\n\nChange-Id: I9b5f02daaf91ecd4416f032834c4fafdcb003b36\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f6d309eed6f0bacce3393553e04265201c2cac55"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f6d309eed6f0bacce3393553e04265201c2cac55"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
