)]}'
{"id":"openstack%2Fcyborg-specs~723786","triplet_id":"openstack%2Fcyborg-specs~master~Icf9f330f04656bf26141b7a658c00fc0382851e7","project":"openstack/cyborg-specs","branch":"master","topic":"add-implemented-directory","hashtags":[],"change_id":"Icf9f330f04656bf26141b7a658c00fc0382851e7","subject":"Add implemented directory","status":"MERGED","created":"2020-04-28 03:44:23.000000000","updated":"2020-09-14 11:48:53.000000000","submitted":"2020-09-14 11:46:27.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"723786-1600083988199-87da3b05","meta_rev_id":"7e1e8db5dd3712c31f39ef0afaa4e09fb25c51ff","_number":723786,"virtual_id_number":723786,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"date":"2020-09-02 08:10:34.000000000","_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":24330,"name":"shaofeng cheng","email":"chengsf@winhong.com","username":"shaofeng_cheng"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-14 11:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":2,"date":"2020-09-14 11:40:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":24330,"name":"shaofeng cheng","email":"chengsf@winhong.com","username":"shaofeng_cheng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":1,"date":"2020-09-14 11:40:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":24330,"name":"shaofeng cheng","email":"chengsf@winhong.com","username":"shaofeng_cheng"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"},{"_account_id":24330,"name":"shaofeng cheng","email":"chengsf@winhong.com","username":"shaofeng_cheng"},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-29 07:49:12.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-08-29 07:49:16.000000000","updated_by":{"_account_id":24330,"name":"shaofeng cheng","email":"chengsf@winhong.com","username":"shaofeng_cheng"},"reviewer":{"_account_id":24330,"name":"shaofeng cheng","email":"chengsf@winhong.com","username":"shaofeng_cheng"},"state":"REVIEWER"},{"updated":"2020-08-29 07:49:22.000000000","updated_by":{"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"},"reviewer":{"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"},"state":"REVIEWER"},{"updated":"2020-08-29 07:49:26.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-08-29 07:50:52.000000000","updated_by":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"reviewer":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"state":"REVIEWER"},{"updated":"2020-09-14 11:40:42.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-09-14 11:46:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d903232f4ad9f2a8c9fe406813ff3da67b94269c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-28 03:44:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c3279ae37ba4a87facbbdb0ebf5bc46d1c937d98","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-28 03:45:06.000000000","message":"Topic changed from victoria-templates to add-implemented-directory","accounts_in_message":[],"_revision_number":1},{"id":"db37b351435878c7d11f23f5fc245f1b1ccb199f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-04-28 03:49:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cce706559ade4ccca3c9eaed169d989e : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e56217dc19224277b27bc7bba95cdc40 : SUCCESS in 3m 35s","accounts_in_message":[],"_revision_number":1},{"id":"74b8585eb9b2f3696168efe9bff1835843594a8d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-04-28 10:49:13.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"0d2a3d2dbe99e967e9485a17f007f896979f17dc","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-28 09:34:28.000000000","message":"Patch Set 1:\n\nrecheck","accounts_in_message":[],"_revision_number":1},{"id":"1d5de21c4306c684ee6332e689ccc49244dd8c39","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-28 09:49:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96dfd341e1364934897237a3cf579642 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e3acb8e634840749815799811333d27 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":1},{"id":"7a7868a37e5f0dc10d98a29667c496d9d2213e22","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-29 07:01:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"91245330d4b59c28e9540ffa74b0674bd9c85e6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 07:06:56.000000000","message":"Patch Set 2: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2574f7c9d84a4ba182cde78d1c220427 : FAILURE in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b7bded7a95a4d17a7f5f5909bf9222a : SUCCESS in 3m 38s","accounts_in_message":[],"_revision_number":2},{"id":"c504ba8646e881af32b9e7c5ade0ad8d9a428360","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-29 07:47:34.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6d5e1e46ac01bbf81cfef9f7c120d87e6aa02045","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 07:51:57.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/244da055950b43fcbb4071db45b8b4d1 : SUCCESS in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/070f2fa7971047978ac7bccba59b38eb : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":3},{"id":"1e40eaeafb02204dc256e939889e2a22346774b5","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 06:52:26.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"c6ccacb3472a0c4c26a18c525ad667a65bbec7b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 07:05:45.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1dfa86f3997146c5b492ec8893cc96fe : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b854610ccd343ac99261fe9d2ab7f50 : SUCCESS in 4m 46s","accounts_in_message":[],"_revision_number":4},{"id":"770b694226c2f7818b862bd7122bb7e88b5ac26d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 07:20:34.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"95e7bae15a8855c538d469451081e470b7b201f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 07:38:10.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/22ec9b28557e43b2b1230a65e92c6ac6 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c8f33d7df964c17bddd5be850736d2e : SUCCESS in 4m 33s","accounts_in_message":[],"_revision_number":5},{"id":"8b1426b9f5345db33626db00c26c1cb495d6fcc8","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 07:58:17.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"0e4d44378a9e9afbd0e5f200443d5c0eaf36d805","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 08:05:28.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2253603eb2a94f0eb6fe56d5fc7f01f3 : SUCCESS in 5m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/662a3048c9e64baf8d6c24d967706f30 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":6},{"id":"466d40487b6b099629d1def66c4572b192b69c44","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 08:10:34.000000000","message":"Patch Set 6:\n\nThis is the temporary link of this change with specs page.\n\nhttps://be08be4bca8d9efed097-90b48a3106a793a661c7495bcec7bb56.ssl.cf2.rackcdn.com/723786/6/check/openstack-tox-docs/2253603/docs/","accounts_in_message":[],"_revision_number":6},{"id":"5263e1a7f329072f5cf51a549e9cac70c85ee77a","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-14 11:40:42.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"f2bb14770203c81e667d05d7fe162ef664683e02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:40:57.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"fba9040299904b00c52b191bfda978142a508355","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:46:27.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0439c4165f5a4077a6613f8244aff940 : SUCCESS in 4m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d5ed55edc7e47cd854a0da3a6a7bfcb : SUCCESS in 4m 19s","accounts_in_message":[],"_revision_number":6},{"id":"40ab70ea798829c6e0ea246efcd2d7e32ce72f17","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:46:28.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"7e1e8db5dd3712c31f39ef0afaa4e09fb25c51ff","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 11:48:53.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/5bc4bd62b23b487d8897dfd9b69cb056 : SUCCESS in 1m 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"83498e9fca29db9e91b5722fc26d3976b01eced8","revisions":{"6eb479c30a3dffafa6e28d31b828bd5c37cfec15":{"kind":"REWORK","_number":1,"created":"2020-04-28 03:44:23.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/86/723786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/86/723786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/1"}}},"commit":{"parents":[{"commit":"f5de14afa31bd15e9f62dc3fdf28d670c8d143c1","subject":"Add spec infrastructure for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f5de14afa31bd15e9f62dc3fdf28d670c8d143c1"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-28 03:40:18.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-28 03:40:22.000000000","tz":480},"subject":"Add implemented directory","message":"Add implemented directory\n\nThe implemented directory is introduced under each release directory to\ndistinguish whether the SPEC approved in the release is completed.\n\nIf the feature is completed within the release cycle, move SPEC from the\napproved directory to the implemented directory\n\nChange-Id: Icf9f330f04656bf26141b7a658c00fc0382851e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6eb479c30a3dffafa6e28d31b828bd5c37cfec15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6eb479c30a3dffafa6e28d31b828bd5c37cfec15"}]},"branch":"refs/heads/master"},"84824a820aeb3bc23e11914a772ee36494157bb1":{"kind":"REWORK","_number":2,"created":"2020-08-29 07:01:33.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/86/723786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/86/723786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/2"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-28 03:40:18.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-08-29 07:00:57.000000000","tz":480},"subject":"Add implemented directory","message":"Add implemented directory\n\nThe implemented directory is introduced under each release directory to\ndistinguish whether the SPEC approved in the release is completed.\n\nIf the feature is completed within the release cycle, move SPEC from the\napproved directory to the implemented directory\n\nChange-Id: Icf9f330f04656bf26141b7a658c00fc0382851e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/84824a820aeb3bc23e11914a772ee36494157bb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/84824a820aeb3bc23e11914a772ee36494157bb1"}]},"branch":"refs/heads/master"},"b9ca10c61fc52ebdea02d0a773a962cb01fda65c":{"kind":"REWORK","_number":3,"created":"2020-08-29 07:47:34.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/86/723786/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/86/723786/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/3"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-28 03:40:18.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-08-29 07:46:43.000000000","tz":480},"subject":"Add implemented directory","message":"Add implemented directory\n\nThe implemented directory is introduced under each release directory to\ndistinguish whether the SPEC approved in the release is completed.\n\nIf the feature is completed within the release cycle, move SPEC from the\napproved directory to the implemented directory\n\nChange-Id: Icf9f330f04656bf26141b7a658c00fc0382851e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b9ca10c61fc52ebdea02d0a773a962cb01fda65c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b9ca10c61fc52ebdea02d0a773a962cb01fda65c"}]},"branch":"refs/heads/master"},"f4fdf0c9aa275df022d69cc862298d27729e1589":{"kind":"REWORK","_number":4,"created":"2020-09-02 06:52:26.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/86/723786/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/86/723786/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/4"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-28 03:40:18.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-09-02 06:52:03.000000000","tz":480},"subject":"Add implemented directory","message":"Add implemented directory\n\nThe implemented directory is introduced under each release directory to\ndistinguish whether the SPEC approved in the release is completed.\n\nIf the feature is completed within the release cycle, move SPEC from the\napproved directory to the implemented directory\n\nChange-Id: Icf9f330f04656bf26141b7a658c00fc0382851e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f4fdf0c9aa275df022d69cc862298d27729e1589"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f4fdf0c9aa275df022d69cc862298d27729e1589"}]},"branch":"refs/heads/master"},"250303e954cb8cf9d3de668ffbd31652e164da9e":{"kind":"REWORK","_number":5,"created":"2020-09-02 07:20:34.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/86/723786/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/86/723786/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/5"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-28 03:40:18.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-09-02 07:18:52.000000000","tz":480},"subject":"Add implemented directory","message":"Add implemented directory\n\nThe implemented directory is introduced under each release directory to\ndistinguish whether the SPEC approved in the release is completed.\n\nIf the feature is completed within the release cycle, move SPEC from the\napproved directory to the implemented directory.\n\nAnother thing is rename the title of every release template.\n\nChange-Id: Icf9f330f04656bf26141b7a658c00fc0382851e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/250303e954cb8cf9d3de668ffbd31652e164da9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/250303e954cb8cf9d3de668ffbd31652e164da9e"}]},"branch":"refs/heads/master"},"83498e9fca29db9e91b5722fc26d3976b01eced8":{"kind":"REWORK","_number":6,"created":"2020-09-02 07:58:17.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/86/723786/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/86/723786/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/86/723786/6"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-04-28 03:40:18.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2020-09-02 07:58:00.000000000","tz":480},"subject":"Add implemented directory","message":"Add implemented directory\n\nThe implemented directory is introduced under each release directory to\ndistinguish whether the SPEC approved in the release is completed.\n\nIf the feature is completed within the release cycle, move SPEC from the\napproved directory to the implemented directory.\n\nAnother thing is rename the title of every release template.\n\nChange-Id: Icf9f330f04656bf26141b7a658c00fc0382851e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/83498e9fca29db9e91b5722fc26d3976b01eced8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/83498e9fca29db9e91b5722fc26d3976b01eced8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
