)]}'
{"id":"openstack%2Fcyborg-specs~728014","triplet_id":"openstack%2Fcyborg-specs~master~I5ff8f67dd577fc6b69e6648ce79d772a5a3ff3e0","project":"openstack/cyborg-specs","branch":"master","topic":"qat","hashtags":[],"change_id":"I5ff8f67dd577fc6b69e6648ce79d772a5a3ff3e0","subject":"Add Intel QAT driver specification.","status":"MERGED","created":"2020-05-14 10:20:28.000000000","updated":"2020-07-31 07:40:38.000000000","submitted":"2020-07-31 07:39:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"728014-1596181158182-3ec634e1","meta_rev_id":"46a989a91a74c87b1a029c30e3fd562a5a7e7b24","_number":728014,"virtual_id_number":728014,"owner":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"date":"2020-07-23 10:35:27.000000000","_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-07-31 07:39:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":2,"date":"2020-07-31 07:30:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":2,"date":"2020-07-28 05:48:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2020-07-31 07:30:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-05-26 11:13:16.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-07-28 05:48:17.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-07-31 07:30:36.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-07-31 07:39:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0f5e0d3daedc943213ce852989d91829a4329f94","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-05-14 10:20:28.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eb0054316b51db0676f10759f47b986d2702a370","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-05-14 10:30:06.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/664f4fc505ba4baabc7d0deb95e1fc9d : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55485eab2895438093a047eb6f6fce6b : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":1},{"id":"7fcc3d7c3120367f4a0694733dac5ecb153a1dc2","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-05-26 11:13:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"313a5a56af5e16c8f654231a8f6cfd6d1045a892","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-06-08 12:03:01.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d8eaaaa764bec91598ffcb50926750bba8a89495","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-07-16 02:58:29.000000000","message":"Patch Set 2: Published edit on patch set 1.","accounts_in_message":[],"_revision_number":2},{"id":"b65615386461661acf5cb840603692072090eac0","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-07-16 03:00:37.000000000","message":"Patch Set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"820633619aea6842179fe27743666fc8baaa977e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-16 03:05:17.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90033eab139e4250855d878182ae3905 : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1537693803f54e52b270b6eee76f02f0 : SUCCESS in 3m 42s","accounts_in_message":[],"_revision_number":3},{"id":"93999a34b2514f64c60af077010187179f953780","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-23 06:10:38.000000000","message":"Patch Set 3:\n\n(2 comments)\n\nLooks good to me overall, but some questions inline.","accounts_in_message":[],"_revision_number":3},{"id":"58dcbcb66cd9ac8fc1b570335273cb77d8fb5dfa","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-07-23 10:35:17.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"38e9af38763e89d2a1806a68c92ab173c5579fbc","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-07-23 10:35:27.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3ef2e399ae79eeb6122b5cde4bc0fab99ec8360e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-23 10:39:46.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f15c380f48f54fe2bc0958f915078b14 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0af593a20ef6481095fdc0dc397c05aa : SUCCESS in 4m 00s","accounts_in_message":[],"_revision_number":4},{"id":"64e7dcd3325956a6da95dfd5a6f748f007ac2f5a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-28 05:48:17.000000000","message":"Patch Set 4: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":4},{"id":"77366f10fc28b2372d73ffa8ceb1259b93db981b","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-07-31 07:30:36.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"eb7645616be9dcae058cbef49e59f3ce6f4c56b9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 07:30:47.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"dcfd95cffed14aeba3dbe81c9e5030ca84ec8381","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 07:39:17.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/48a22741dc734ae1b8a239a1a5697f39 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cfadb27b12864cafa77099994b5ed8bf : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":4},{"id":"6efea4305950fb5b91e83bb1b6d90c9da620e339","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 07:39:18.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":4},{"id":"fa1d926918d388d1ad8eac7613bdac7ca50ecd07","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-31 07:40:38.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/94a3a66046484c0e8fad10b125d4a82d : SUCCESS in 59s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"7244d2f4b413c36ace269a1f299523ef49ea71af","revisions":{"1811a728939b910465e4637f6401717ff43265e8":{"kind":"REWORK","_number":1,"created":"2020-05-14 10:20:28.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/14/728014/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/14/728014/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/1"}}},"commit":{"parents":[{"commit":"f5de14afa31bd15e9f62dc3fdf28d670c8d143c1","subject":"Add spec infrastructure for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f5de14afa31bd15e9f62dc3fdf28d670c8d143c1"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-05-14 10:16:09.000000000","tz":480},"committer":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-05-14 10:17:38.000000000","tz":480},"subject":"Add Intel QAT driver specification.","message":"Add Intel QAT driver specification.\n\nChange-Id: I5ff8f67dd577fc6b69e6648ce79d772a5a3ff3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1811a728939b910465e4637f6401717ff43265e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1811a728939b910465e4637f6401717ff43265e8"}]},"branch":"refs/heads/master"},"da524468c08fc8c422da4814f45751240c0eddea":{"kind":"REWORK","_number":2,"created":"2020-07-16 02:58:29.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/14/728014/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/14/728014/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/2"}}},"commit":{"parents":[{"commit":"f5de14afa31bd15e9f62dc3fdf28d670c8d143c1","subject":"Add spec infrastructure for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f5de14afa31bd15e9f62dc3fdf28d670c8d143c1"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-05-14 10:16:09.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2020-07-16 02:57:09.000000000","tz":0},"subject":"Add Intel QAT driver specification.","message":"Add Intel QAT driver specification.\n\nStroy: 2007773\nChange-Id: I5ff8f67dd577fc6b69e6648ce79d772a5a3ff3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/da524468c08fc8c422da4814f45751240c0eddea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/da524468c08fc8c422da4814f45751240c0eddea"}]},"branch":"refs/heads/master"},"e73767ad4cef07fa9fb11434aeafe7ddde71e016":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-07-16 03:00:37.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/14/728014/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/14/728014/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/3"}}},"commit":{"parents":[{"commit":"f5de14afa31bd15e9f62dc3fdf28d670c8d143c1","subject":"Add spec infrastructure for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f5de14afa31bd15e9f62dc3fdf28d670c8d143c1"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-05-14 10:16:09.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2020-07-16 03:00:30.000000000","tz":0},"subject":"Add Intel QAT driver specification.","message":"Add Intel QAT driver specification.\n\nStroy: 2007773\nTask: 39998\n\nChange-Id: I5ff8f67dd577fc6b69e6648ce79d772a5a3ff3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e73767ad4cef07fa9fb11434aeafe7ddde71e016"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e73767ad4cef07fa9fb11434aeafe7ddde71e016"}]},"branch":"refs/heads/master"},"7244d2f4b413c36ace269a1f299523ef49ea71af":{"kind":"REWORK","_number":4,"created":"2020-07-23 10:35:17.000000000","uploader":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"ref":"refs/changes/14/728014/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/14/728014/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/14/728014/4"}}},"commit":{"parents":[{"commit":"f5de14afa31bd15e9f62dc3fdf28d670c8d143c1","subject":"Add spec infrastructure for Victoria","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/f5de14afa31bd15e9f62dc3fdf28d670c8d143c1"}]}],"author":{"name":"Xinran Wang","email":"xin-ran.wang@intel.com","date":"2020-05-14 10:16:09.000000000","tz":480},"committer":{"name":"Xinran WANG","email":"xin-ran.wang@intel.com","date":"2020-07-23 10:35:12.000000000","tz":0},"subject":"Add Intel QAT driver specification.","message":"Add Intel QAT driver specification.\n\nStroy: 2007773\nTask: 39998\n\nChange-Id: I5ff8f67dd577fc6b69e6648ce79d772a5a3ff3e0\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/7244d2f4b413c36ace269a1f299523ef49ea71af"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/7244d2f4b413c36ace269a1f299523ef49ea71af"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
