)]}'
{"id":"openstack%2Fcyborg-specs~738331","triplet_id":"openstack%2Fcyborg-specs~master~I72acf569d56d4e7feaec83da9274364282cf0dbf","project":"openstack/cyborg-specs","branch":"master","topic":"attribute-api-support","hashtags":[],"change_id":"I72acf569d56d4e7feaec83da9274364282cf0dbf","subject":"Add Attribute API specification.","status":"ABANDONED","created":"2020-06-28 02:15:18.000000000","updated":"2022-05-27 06:22:22.000000000","total_comment_count":70,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"5301b5fcfa09e8f07e5714564c650ca91ad18225","_number":738331,"virtual_id_number":738331,"owner":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-09-17 02:22:52.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":1,"date":"2020-09-22 08:19:07.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":-1,"date":"2020-11-17 08:28:28.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-09-25 01:03:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-06-28 07:36:48.000000000","updated_by":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"reviewer":{"_account_id":23168,"name":"Yumeng Bao","username":"Yumeng","inactive":true},"state":"REVIEWER"},{"updated":"2020-06-28 10:19:33.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2020-08-25 08:42:09.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-09-17 02:22:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-09-22 08:19:07.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-09-25 01:03:24.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-11-17 08:28:28.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"5bbd489c730974d781974f4da16b3c895c3b813c","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-06-28 02:15:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"984b39830579eb0e40ea66def9bb910e0786d37d","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-06-28 02:17:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"964419de08bb48c6932219966939a50338549a89","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-06-28 02:19:53.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e40594eb3e16800e9e1b6f1346a61e20424a9923","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 02:27:59.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8917147248ec40b5b0052a0dc11652a9 : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ceab63d16a9e4370a3629617ed78f44f : FAILURE in 3m 06s","accounts_in_message":[],"_revision_number":3},{"id":"1217d61c6f40b24de0d1fe42dce2a5ecdaaa214b","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-06-28 04:28:03.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"b716aa0a688789ca182dbffbe307a3b9d5e77a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-06-28 04:32:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb89035ca24e4e5e903450a8fb21721e : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f6ae8fd10bc14ba6933fb85cf1d032de : SUCCESS in 3m 50s","accounts_in_message":[],"_revision_number":4},{"id":"3e30107c12716b82b7e240bd569193c991e7e53b","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-06-28 09:32:45.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nThanks zhenghao.","accounts_in_message":[],"_revision_number":4},{"id":"caa2d1305ef02fd9c65c440e4ef9d13b5516c43a","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-06-28 09:50:48.000000000","message":"Patch Set 4: Code-Review-1\n\n(5 comments)\n\nThanks zhenghao. Please see my comments inline.","accounts_in_message":[],"_revision_number":4},{"id":"b4097adc8c29a74eb14ad4e956d353226f436936","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-06-28 10:19:33.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"7d53a16f12b8381dbdd5514bd225fe8317e8f083","author":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"date":"2020-06-28 16:26:13.000000000","message":"Patch Set 4:\n\nsorry, typo\n\nAt present, only admin allow to do program, so this maybe a program(s/program/problem).\nOne solution is that, he can write a script, list all the expected FPGA devices and set attribute for them.\n\nBut this maybe not a program(s/program/problem) for tenant program, for tenant own less devices.","accounts_in_message":[],"_revision_number":4},{"id":"ed9aaf8857d1e8fd615f24facc559a845f66ee0b","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-07-01 13:50:07.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"8e4d3b5ede44e1e9dd9c3e1fb3a6b7a2b1ddb323","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-07-01 13:50:40.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"ff8a562921b769ad8c2bbf8a457f49b9fb3e6935","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-01 13:57:50.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6c2ed9a824304c9fb033a79032d69688 : SUCCESS in 6m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/112e01f3c02c4304b28169cc0a47c900 : SUCCESS in 5m 54s","accounts_in_message":[],"_revision_number":5},{"id":"f1fadcd95088c21e0343f2b7f61a1eaac6ccfc5c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-02 00:50:20.000000000","message":"Patch Set 5: Code-Review-1\n\n(15 comments)","accounts_in_message":[],"_revision_number":5},{"id":"27fbea6d1d126bca48b3173205b6ab140374e8a3","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-07-08 16:36:54.000000000","message":"Patch Set 5:\n\n(8 comments)","accounts_in_message":[],"_revision_number":5},{"id":"dc0e0d1a7867ff9a26c454f2c5057ce78b702472","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-07-08 16:38:52.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"c6bd4ea0ba21d12933f05e49063f4159aba1e213","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-08 16:47:40.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3acb5c9b40e44538f97c867774dac2e : FAILURE in 5m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20330f28a86742f0a283d4da7635b085 : SUCCESS in 6m 36s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/victoria/approved/attribute-api-support.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/victoria/approved/attribute-api-support.rst","accounts_in_message":[],"_revision_number":6},{"id":"f529443331c3a509e4f135699ab6bd11ae70fa81","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-07-14 11:56:48.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"2296ca01fbbf4f9ac88fd5ec18650d2ba399039d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-07-14 12:00:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eea216f9ad444ba3a0b272c2918e7d31 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1abca3dfd8f9441584b2b3c7cbbb4895 : SUCCESS in 3m 32s","accounts_in_message":[],"_revision_number":7},{"id":"63be33d09d4dbe37533ba6b5550f1440089ffc19","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-07-16 03:30:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"41df5331fcf34760e13b85800d63c274d0cf1555","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-07-16 07:16:27.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"0655c28bb0b02b1d86ff8f95d605d920a3bac11a","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-07-16 07:56:44.000000000","message":"Patch Set 7: Code-Review-1\n\n(1 comment)\n\nReference Cyborg API Version 2 docs, hope you can improve the current spec, the docs import to us and to the users.\nhttps://specs.openstack.org/openstack/cyborg-specs/specs/ussuri/approved/cyborg-api.html","accounts_in_message":[],"_revision_number":7},{"id":"bae1c5e3983130998ace5aa0a64b8e77491949e7","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-08-25 02:09:39.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5638750f250fa9d02655e3ec425b8db7986425b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 02:18:05.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e92b3e734f6e43169db4efee01131fc2 : SUCCESS in 3m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8b9e053092a4481b6cc8e8dd094058a : SUCCESS in 7m 57s","accounts_in_message":[],"_revision_number":8},{"id":"4c6e51799d67436107dcc93a0b3fc8f225501a72","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-25 06:47:32.000000000","message":"Patch Set 8:\n\n(6 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0228b8f1edc3ae4deba06f4074195288e09a1eae","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-25 06:47:39.000000000","message":"Patch Set 9: Published edit on patch set 8.","accounts_in_message":[],"_revision_number":9},{"id":"024e88e510fcee2bd0db5c0942aa3f8e20ee9fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 07:05:37.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04f3582449024716b2f41ac26c0960da : SUCCESS in 5m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e51e0713281f42148cfeb0bb08921a78 : SUCCESS in 4m 57s","accounts_in_message":[],"_revision_number":9},{"id":"e5f6015ec2a3fcd68b364b070bd189ee220f3fe1","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-08-25 07:19:47.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"ab772cbdb681e4a790452204f8d77a762caa2ab0","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-08-25 07:20:48.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"88e786ed7a1aed94f9acb78a4851a969d345845e","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-08-25 07:27:10.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"7314228c0687ef4f37b5fa330b4d7c3a19967add","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-25 07:51:45.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"6f29cb73745fb1bed42512a09b545f85140f1178","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 08:15:05.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/87483c5bfd4d4fd7ad2ce75ef050d704 : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8be2949f9d04b4cb1a119da59fe7d4e : SUCCESS in 4m 02s","accounts_in_message":[],"_revision_number":11},{"id":"0dd96cce5ee57ae30c4ecdd5a10df268945a893b","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-08-25 08:28:12.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"245e14620d206157a556f63cac617a190198735a","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-08-25 08:30:54.000000000","message":"Patch Set 12:\n\n(2 comments)\n\n@Wenping Thx for your review. Plz look at the reply in line and I have add batch delete.","accounts_in_message":[],"_revision_number":12},{"id":"cb4b48517ad5e867013558bb477066ee91f4fc2e","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-25 08:42:09.000000000","message":"Patch Set 12:\n\nLGTM. Thanks for quickly reply.","accounts_in_message":[],"_revision_number":12},{"id":"d7481b71a3f6cdff222f8ba17c93300d8065f071","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-25 08:43:46.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9bd267138db46bda581f0694b306c9a : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/795a0cb12e7f4aad8dfe2a8c91eaefda : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":12},{"id":"52bcf02475e5cd43b138c1e0c88d27d7b5d18b5c","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-08-25 10:15:34.000000000","message":"Patch Set 12: Code-Review-1\n\nI am sorry wangzhih, the SPEC of Victoria release was frozen, we should move this feature to wallaby directory, and we can approve this in Wallaby release.\n\nhttps://review.opendev.org/#/c/747888/ is add the wallaby directory, that you can rebased it on that patch.","accounts_in_message":[],"_revision_number":12},{"id":"736a0dcb502e24309af8afad981358f7eb624fd6","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-09-02 03:42:47.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"928fce208432736a78f298e78bac472a27509305","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 03:49:04.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3cb5dbb935174e9fafd622e0ac082dd2 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dd65db7ffa214a0182d96f9b5ee16248 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":13},{"id":"78e334d2c3a06bd818bb9dc55eb42ff028756755","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-15 10:48:48.000000000","message":"Patch Set 13: Code-Review-1\n\n(7 comments)\n\nThanks for your updates, please see my comments inline. Please let me know if I misunderstanding something. :)","accounts_in_message":[],"_revision_number":13},{"id":"eaeea6f58ffcff8627b76098b6543e71dc8bdf53","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-09-15 11:33:45.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"1ed86c99fcd168eef7de128ce66fa3a03f3f528e","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-09-15 11:36:35.000000000","message":"Patch Set 14:\n\n(4 comments)\n\nHi xinran, thx for your comment. Plz review again. Thx.","accounts_in_message":[],"_revision_number":14},{"id":"b44412df0cfd3c9ecb67c133c4b7dd3cbd526052","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 11:39:08.000000000","message":"Patch Set 14: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dcaed8bd61e743f49a5cb925badaddf4 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f83ad9340f5d47d9bfa86c84aa01c0fd : POST_FAILURE in 3m 26s","accounts_in_message":[],"_revision_number":14},{"id":"f9198823bce6635e74b2b2b725b205c9505f2755","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-09-15 11:55:40.000000000","message":"Patch Set 14:\n\nrecheck","accounts_in_message":[],"_revision_number":14},{"id":"7f0b4ddad59de3b5595c243fdf365be465447119","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 12:01:02.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a5407d06ea244c0b19b7e8309e93d00 : SUCCESS in 4m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9d4b6f477a12431e94ea6792938bf4ff : SUCCESS in 3m 15s","accounts_in_message":[],"_revision_number":14},{"id":"7346c6216cef7afaa28a46a1e6089ebaee1da8f1","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-15 12:44:22.000000000","message":"Patch Set 14:\n\n(3 comments)\n\nHi zhenghai, overal looks good to me.\nsoft -1 for 404 situations. Please see my comments inline.","accounts_in_message":[],"_revision_number":14},{"id":"5d4c820f76cfc57c9a1206360015890c3c41434b","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-15 12:44:26.000000000","message":"Patch Set 14: Code-Review-1","accounts_in_message":[],"_revision_number":14},{"id":"af74b5996c61ec24e3a6349981c03baf3cf8b016","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-15 12:54:11.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"4069c8efc23595fd63b376e3c4ce45550c393fe8","author":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"date":"2020-09-15 14:06:20.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"e5249e05a192f7540ad175767337d010c8ac0858","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 14:12:09.000000000","message":"Patch Set 15: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e13a9996444429a551162189b045cf : POST_FAILURE in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83c4196a225644fdb771c0a56f26f8ff : SUCCESS in 3m 25s","accounts_in_message":[],"_revision_number":15},{"id":"1c1ba3d70d20e6f726361bf8042007c4349da621","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-17 02:13:23.000000000","message":"Patch Set 15:\n\nrecheck","accounts_in_message":[],"_revision_number":15},{"id":"601ac610ddcb7bda51dff28ef5075d006efc2014","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-17 02:22:52.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/84c5e08a4c1f4fdf86eb3489462309bb : SUCCESS in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e4518e5e8bc42fd857c027561bfe3ca : SUCCESS in 2m 47s","accounts_in_message":[],"_revision_number":15},{"id":"4ee4d894dfe8df6eb628b861e83f1eed6621eb38","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-09-22 08:19:07.000000000","message":"Patch Set 15: Code-Review+1\n\nhi zheng hao, overall looks good, but one reminder,\nplease consider policy for each attribute APIs and update here https://wiki.openstack.org/wiki/Cyborg/Policy","accounts_in_message":[],"_revision_number":15},{"id":"edb6211acb8dc2ba3fbd7ca5139db308631fd4bb","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-25 01:03:24.000000000","message":"Patch Set 15: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":15},{"id":"0276df14d46847dd0d92ccc32b846e91af42d560","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-17 08:28:28.000000000","message":"Patch Set 15: Code-Review-1\n\n(2 comments)\n\nLGTM, but one doubt with * Change Cyborg Attribute table.*.","accounts_in_message":[],"_revision_number":15},{"id":"5301b5fcfa09e8f07e5714564c650ca91ad18225","tag":"autogenerated:gerrit:abandon","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-05-27 06:22:22.000000000","message":"Abandoned\n\nThere is no long update for two release, if you will continue to work for this, please re-propose it in latest version.","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"2408635f825360642da6098e5108f81912f05c4a","revisions":{"357ab4521c6248daea8ce94e33506ea0d336d618":{"kind":"REWORK","_number":1,"created":"2020-06-28 02:15:18.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/1"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:14:50.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/357ab4521c6248daea8ce94e33506ea0d336d618"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/357ab4521c6248daea8ce94e33506ea0d336d618"}]},"branch":"refs/heads/master"},"0b35543832ebe3f50c6d05f86c25605c795ead21":{"kind":"REWORK","_number":2,"created":"2020-06-28 02:17:18.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/2"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:16:56.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0b35543832ebe3f50c6d05f86c25605c795ead21"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0b35543832ebe3f50c6d05f86c25605c795ead21"}]},"branch":"refs/heads/master"},"b17f0220fac771ccd7bed69b3cafcf30b85422b0":{"kind":"REWORK","_number":3,"created":"2020-06-28 02:19:53.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/3"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:19:33.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b17f0220fac771ccd7bed69b3cafcf30b85422b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b17f0220fac771ccd7bed69b3cafcf30b85422b0"}]},"branch":"refs/heads/master"},"205c9d539ad0d3bd67e8610aced1a16a72143238":{"kind":"REWORK","_number":4,"created":"2020-06-28 04:28:03.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/4"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 04:27:36.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/205c9d539ad0d3bd67e8610aced1a16a72143238"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/205c9d539ad0d3bd67e8610aced1a16a72143238"}]},"branch":"refs/heads/master"},"be72151fd3a1f84a3c1a8f3f0fc0b0452ce7cea6":{"kind":"REWORK","_number":5,"created":"2020-07-01 13:50:07.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/5"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-07-01 13:49:32.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/be72151fd3a1f84a3c1a8f3f0fc0b0452ce7cea6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/be72151fd3a1f84a3c1a8f3f0fc0b0452ce7cea6"}]},"branch":"refs/heads/master"},"235bf151de1842e8e8d9b5751ed66f9069a83013":{"kind":"REWORK","_number":6,"created":"2020-07-08 16:38:52.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/6"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-07-08 16:38:03.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/235bf151de1842e8e8d9b5751ed66f9069a83013"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/235bf151de1842e8e8d9b5751ed66f9069a83013"}]},"branch":"refs/heads/master"},"76805f90ae6a12ffd1b6eac81f085c3383a0df39":{"kind":"REWORK","_number":7,"created":"2020-07-14 11:56:48.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/7"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhh","email":"wangzhengh@chinatelecom.cn","date":"2020-07-14 11:52:30.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/76805f90ae6a12ffd1b6eac81f085c3383a0df39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/76805f90ae6a12ffd1b6eac81f085c3383a0df39"}]},"branch":"refs/heads/master"},"13cf321e553f89862cd63b2d1492cb390cc11ca2":{"kind":"REWORK","_number":8,"created":"2020-08-25 02:09:39.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/8"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhh","email":"wangzhengh@chinatelecom.cn","date":"2020-08-25 02:07:15.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/13cf321e553f89862cd63b2d1492cb390cc11ca2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/13cf321e553f89862cd63b2d1492cb390cc11ca2"}]},"branch":"refs/heads/master"},"56903f217b7f7faa71ed28bc7f19ce07295bea78":{"kind":"REWORK","_number":9,"created":"2020-08-25 06:47:39.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/31/738331/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/9"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-08-25 06:41:02.000000000","tz":0},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/56903f217b7f7faa71ed28bc7f19ce07295bea78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/56903f217b7f7faa71ed28bc7f19ce07295bea78"}]},"branch":"refs/heads/master"},"7b37ce7dba2d2a5b7fad681a5e3bf62972db84a8":{"kind":"REWORK","_number":10,"created":"2020-08-25 07:19:47.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/10"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhh","email":"wangzhengh@chinatelecom.cn","date":"2020-08-25 07:18:45.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/7b37ce7dba2d2a5b7fad681a5e3bf62972db84a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/7b37ce7dba2d2a5b7fad681a5e3bf62972db84a8"}]},"branch":"refs/heads/master"},"8843c8e9e68f788751a83d5b227fc40bb2aa874c":{"kind":"REWORK","_number":11,"created":"2020-08-25 07:27:10.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/11"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhh","email":"wangzhengh@chinatelecom.cn","date":"2020-08-25 07:26:23.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/8843c8e9e68f788751a83d5b227fc40bb2aa874c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/8843c8e9e68f788751a83d5b227fc40bb2aa874c"}]},"branch":"refs/heads/master"},"bf329d4c14ebf870581bfd4125c1cc25e0007cc3":{"kind":"REWORK","_number":12,"created":"2020-08-25 08:28:12.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/12"}}},"commit":{"parents":[{"commit":"0a718d426dbf79eafb21f03f981b08f93f185962","subject":"Remove Babel","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a718d426dbf79eafb21f03f981b08f93f185962"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhh","email":"wangzhengh@chinatelecom.cn","date":"2020-08-25 08:27:50.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/bf329d4c14ebf870581bfd4125c1cc25e0007cc3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/bf329d4c14ebf870581bfd4125c1cc25e0007cc3"}]},"branch":"refs/heads/master"},"21599d979e8ffc5e074fb8bc350b2e93d25ff430":{"kind":"REWORK","_number":13,"created":"2020-09-02 03:42:47.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/13"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhh","email":"wangzhengh@chinatelecom.cn","date":"2020-09-02 03:39:21.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/21599d979e8ffc5e074fb8bc350b2e93d25ff430"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/21599d979e8ffc5e074fb8bc350b2e93d25ff430"}]},"branch":"refs/heads/master"},"29e107bea20458e365fda26ea56f06e76d727fe4":{"kind":"REWORK","_number":14,"created":"2020-09-15 11:33:45.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/14"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhengh","email":"wangzhengh@chinatelecom.cn","date":"2020-09-15 11:33:27.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/29e107bea20458e365fda26ea56f06e76d727fe4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/29e107bea20458e365fda26ea56f06e76d727fe4"}]},"branch":"refs/heads/master"},"2408635f825360642da6098e5108f81912f05c4a":{"kind":"REWORK","_number":15,"created":"2020-09-15 14:06:20.000000000","uploader":{"_account_id":17813,"name":"wangzhh","email":"wzh_1993@126.com","username":"wangzhh"},"ref":"refs/changes/31/738331/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/31/738331/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/31/738331/15"}}},"commit":{"parents":[{"commit":"5289b226e3b84654414dd2b78216339ac8981687","subject":"Create specs directory for Wallaby","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5289b226e3b84654414dd2b78216339ac8981687"}]}],"author":{"name":"wangzhenghao","email":"wangzhengh@chinatelecom.cn","date":"2020-06-28 02:13:59.000000000","tz":480},"committer":{"name":"wangzhh","email":"wangzhengh@chinatelecom.cn","date":"2020-09-15 14:05:05.000000000","tz":480},"subject":"Add Attribute API specification.","message":"Add Attribute API specification.\n\nChange-Id: I72acf569d56d4e7feaec83da9274364282cf0dbf\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2408635f825360642da6098e5108f81912f05c4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2408635f825360642da6098e5108f81912f05c4a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
