)]}'
{"id":"openstack%2Fcyborg-specs~748804","triplet_id":"openstack%2Fcyborg-specs~master~I576b64f8e0e1116091e51de42e560acfc2fd1fc6","project":"openstack/cyborg-specs","branch":"master","topic":"adjust-nova-cyborg-interaction-specs","hashtags":[],"change_id":"I576b64f8e0e1116091e51de42e560acfc2fd1fc6","subject":"Add yasfb to generate RSS feeds.","status":"MERGED","created":"2020-08-29 10:03:08.000000000","updated":"2020-09-15 04:51:08.000000000","submitted":"2020-09-15 04:48:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"748804-1600145338527-a980bdbc","meta_rev_id":"5ae0fba7293d49d4ec6e5db4dc3511b812259214","_number":748804,"virtual_id_number":748804,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-09-15 04:48:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-09-14 12:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2020-09-14 12:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2020-09-14 12:17:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-08-31 02:56:28.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-08-31 02:56:36.000000000","updated_by":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"reviewer":{"_account_id":14131,"name":"shaohef","email":"shaohe.feng@intel.com","username":"shaohefeng"},"state":"REVIEWER"},{"updated":"2020-08-31 02:56:39.000000000","updated_by":{"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"},"reviewer":{"_account_id":23982,"name":"shogoadachi","email":"sho-adachi@xc.jp.nec.com","username":"ShogoAdachi"},"state":"REVIEWER"},{"updated":"2020-09-04 07:59:29.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-09-04 09:56:37.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-09-14 12:17:46.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2020-09-15 04:48:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac375b4840471598a9a107c5c0838c5c153c9984","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-08-29 10:03:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5bf4c95aee37f447d6ab3cb693b95757ad9b34ee","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-08-29 10:14:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/459d0ca8ee534616a6b2efd6803fa0b0 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8786ead03cab461b8b08893993b458e8 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":1},{"id":"d5056f090d133fe7638ba61693ea3677012ba941","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-02 07:36:14.000000000","message":"Patch Set 2: Patch Set 1 was rebased","accounts_in_message":[],"_revision_number":2},{"id":"104c4cd655f8063040ac31abdb65b07197be7a1c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-02 07:48:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f6b050b87ad4a0488187e8f6af73a88 : SUCCESS in 6m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5a921065e234f13bb2bd536dab39e18 : SUCCESS in 4m 11s","accounts_in_message":[],"_revision_number":2},{"id":"cf5ad9b742fdc0b36e03ac5728f78f3dbdb1f629","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2020-09-04 07:59:29.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"7a0248468638d38a801a8fbb34a1cc5267d51bae","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-09-04 09:56:37.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\nLGTM","accounts_in_message":[],"_revision_number":2},{"id":"37fc71be801d6c28f14d16050091bf43d5b44d45","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-09-14 12:03:26.000000000","message":"Patch Set 3: Patch Set 2 was rebased","accounts_in_message":[],"_revision_number":3},{"id":"b8474454c3a68c649efa7b1cdf99a737925b9c88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-14 12:12:47.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e43afc5923504fd1989051613329abaf : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7067b13a25845e7803cf1fbcc6df4fe : SUCCESS in 2m 57s","accounts_in_message":[],"_revision_number":3},{"id":"cf9b4428f4a2d39b552748db307ec674babbaf20","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-09-14 12:17:46.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"f5e1a354b7cf3da536b0b36ec44954def3689d07","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 04:43:46.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"bb17bb08a215519ab6f4ef3bb469bb2966981422","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 04:48:58.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8c73286c70e542108ccd9f5d20c6f9a0 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0eef50f5fb242148d2c08a944fb19e6 : SUCCESS in 3m 20s","accounts_in_message":[],"_revision_number":3},{"id":"88a9afacf03d8d2e51882b91e548beccf4d267cb","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 04:48:58.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":3},{"id":"24590dff549f1d7895fe2c2eaef901182e1e67e4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-09-15 04:51:08.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/4a3ed824539a48b8815b7dc59ae13b3d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"8c6c1c50a7abbdd9a67093d46fbbe9cd0a9d6f83","revisions":{"ccd164723e40c2334df42acd3fba4d5b36ce7e42":{"kind":"REWORK","_number":1,"created":"2020-08-29 10:03:08.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/04/748804/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/04/748804/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/1"}}},"commit":{"parents":[{"commit":"bbf3bb0e42a7d36b5cb5f6a0243128a7a28b8b38","subject":"Adjust the nova-cyborg-interaction specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/bbf3bb0e42a7d36b5cb5f6a0243128a7a28b8b38"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-08-29 09:58:01.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-08-29 09:58:08.000000000","tz":480},"subject":"Add yasfb to generate RSS feeds.","message":"Add yasfb to generate RSS feeds.\n\nSet up yasfb to generate an RSS feed for approved specs.\nyasfb is used by all spec projects to generate RSS feeds.\n\nChange-Id: I576b64f8e0e1116091e51de42e560acfc2fd1fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ccd164723e40c2334df42acd3fba4d5b36ce7e42"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ccd164723e40c2334df42acd3fba4d5b36ce7e42"}]},"branch":"refs/heads/master"},"b6a2df959c1a1e01acc3805dc7877a8d258af210":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2020-09-02 07:36:14.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/04/748804/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/04/748804/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/2"}}},"commit":{"parents":[{"commit":"ceb2c28c0aaa51b51235a50e5939219c847d7d9b","subject":"Adjust the nova-cyborg-interaction specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ceb2c28c0aaa51b51235a50e5939219c847d7d9b"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-08-29 09:58:01.000000000","tz":480},"committer":{"name":"Brin Zhang","email":"zhangbailin@inspur.com","date":"2020-09-02 07:36:14.000000000","tz":0},"subject":"Add yasfb to generate RSS feeds.","message":"Add yasfb to generate RSS feeds.\n\nSet up yasfb to generate an RSS feed for approved specs.\nyasfb is used by all spec projects to generate RSS feeds.\n\nChange-Id: I576b64f8e0e1116091e51de42e560acfc2fd1fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b6a2df959c1a1e01acc3805dc7877a8d258af210"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b6a2df959c1a1e01acc3805dc7877a8d258af210"}]},"branch":"refs/heads/master"},"8c6c1c50a7abbdd9a67093d46fbbe9cd0a9d6f83":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2020-09-14 12:03:26.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/04/748804/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/04/748804/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/04/748804/3"}}},"commit":{"parents":[{"commit":"8752d94e991d74372533f6a05b432ab46b12d012","subject":"Adjust the nova-cyborg-interaction specs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/8752d94e991d74372533f6a05b432ab46b12d012"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-08-29 09:58:01.000000000","tz":480},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2020-09-14 12:03:26.000000000","tz":0},"subject":"Add yasfb to generate RSS feeds.","message":"Add yasfb to generate RSS feeds.\n\nSet up yasfb to generate an RSS feed for approved specs.\nyasfb is used by all spec projects to generate RSS feeds.\n\nChange-Id: I576b64f8e0e1116091e51de42e560acfc2fd1fc6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/8c6c1c50a7abbdd9a67093d46fbbe9cd0a9d6f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/8c6c1c50a7abbdd9a67093d46fbbe9cd0a9d6f83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
