)]}'
{"id":"openstack%2Fcyborg-specs~758925","triplet_id":"openstack%2Fcyborg-specs~master~If44ad617fcf544690dcbaad1183953c68440bc3b","project":"openstack/cyborg-specs","branch":"master","topic":"enable-vgpu-in-cyborg","hashtags":[],"change_id":"If44ad617fcf544690dcbaad1183953c68440bc3b","subject":"Cyborg NVIDIA GPU Driver support vGPU management","status":"MERGED","created":"2020-10-21 03:14:22.000000000","updated":"2020-11-27 09:11:07.000000000","submitted":"2020-11-27 09:07:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":96,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"758925-enable-vgpu-in-cyborg","meta_rev_id":"4b32fa031757fea673c9fe0fce2a880f6694fd4c","_number":758925,"virtual_id_number":758925,"owner":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2020-11-27 09:07:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2020-11-27 08:38:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2020-11-27 08:28:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2020-11-27 08:58:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-21 03:15:51.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-10-21 03:16:12.000000000","updated_by":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"reviewer":{"_account_id":14107,"name":"zhurong","email":"aaronzhu1121@gmail.com","username":"zhurong"},"state":"REVIEWER"},{"updated":"2020-10-21 03:17:17.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2020-11-17 09:18:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-17 11:34:39.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2020-11-19 06:17:05.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-11-19 06:31:15.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"b55129f0c888661185b63b3ef3af13980c024f55","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-21 03:14:22.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2370694192b7e757843bbe2e84be93cd5adf0030","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 03:21:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dd10bf1553f24f42a57a171a5e7c5052 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd6c33f96a8a4f9fb09774d0814f6678 : SUCCESS in 5m 06s","accounts_in_message":[],"_revision_number":1},{"id":"225b1b9e1cce9eb36e41939fdb5c69e583b63be9","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-21 03:39:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"dbc9ee7e294673140c035db9be3502f3b54bc52a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-21 03:45:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52967693b7c041ffa5c7dcf89205d924 : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3fa201df8de47d590cf8457441f952f : SUCCESS in 4m 25s","accounts_in_message":[],"_revision_number":2},{"id":"74ff04447427b55eee8f452f5710b6f1eba79e75","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-21 07:23:31.000000000","message":"Patch Set 2: Code-Review-1\n\n(12 comments)\n\nGood job, thanks for starting this work.","accounts_in_message":[],"_revision_number":2},{"id":"80442aa71c1fd72dff560e325359526fa85c19d2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-21 07:30:06.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84e3f93ca3035c42ff50166a8a7edb99cfe18709","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-24 03:12:24.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"dc4323a39af3786cbf2b4cfa5a4eb60edd082b32","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-24 03:15:43.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nThanks Brin for the review! Please see my updates in the new patch.","accounts_in_message":[],"_revision_number":1},{"id":"5a625ef62bbb0c398e65c90e2b032791047469c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 03:16:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cfa904c92c14304ab71e9b7b470d538 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff8eab095da042cfad0e00d87dec1986 : SUCCESS in 2m 44s","accounts_in_message":[],"_revision_number":3},{"id":"5495f0bfc3e0f8fc63bad36d6a92d9ce450851df","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-10-24 07:25:48.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"211105a78c15342ae22bcd240e54b563a04536aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 07:29:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd9de964a447440a8fd58da63fe2a014 : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53f89c98ee3473db7879c4f1215dad6 : SUCCESS in 2m 36s","accounts_in_message":[],"_revision_number":4},{"id":"fb871c8d24e949e0edfa9fb6ec6db6de32ecbed7","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-24 08:53:14.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"4a5260be521e9e5cb6df5484d54ab8bc23ce6b29","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-04 02:14:23.000000000","message":"Topic set to enable-vgpu-in-cyborg","accounts_in_message":[],"_revision_number":4},{"id":"18873e6e7c6b6a32018c5d8bbaa8e2aaeab1ca8f","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-05 06:36:10.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"48cafe290a97b8c89285e975c50ccfb621203c4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-05 06:50:28.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9670f1692ea043948a6180f2194c154b : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a50d2f184c3547379d5d83349f1cfcf3 : SUCCESS in 3m 55s","accounts_in_message":[],"_revision_number":5},{"id":"4738cf3e698e534d8e165753870c5f8662e1be4c","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-06 02:26:08.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"aba929175e1c209b6c5e577d0c782fa8c4ba4083","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-06 02:43:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"89de58035bda7d8369c30d686fed8805ffb10e36","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-06 08:49:08.000000000","message":"Patch Set 5: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":5},{"id":"3d322ec31e6812df02f948fc058ea6127c53fc58","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-06 12:28:11.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"674b049955cf82e656048ee855a9cc83ef31a8b3","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-06 12:31:15.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"79aff589fe8a2b9531ad0247f31306639b955b88","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-06 12:33:19.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"64f989f14654acd1684a08bb75e5f1b77d7c8602","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-06 12:45:04.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fc3a466380d4a0d8e8f38fd8deec4d5 : FAILURE in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0a697d62ff94562802f9b19c134dafa : FAILURE in 3m 34s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/wallaby/approved/vgpu-driver-proposal.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/wallaby/approved/vgpu-driver-proposal.rst","accounts_in_message":[],"_revision_number":8},{"id":"32946ecbbe93af05794ff5b14822f1e42a1363f4","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-09 06:58:39.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"fbf1b1286e5762b3e1f39f6d8cd034c4c3ae957e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-09 07:21:03.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecc02b6395494889bc5d493ab943cc80 : SUCCESS in 4m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2bcdc9f7db694bbd85824685586e55f9 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":9},{"id":"520d0ec190d2e8fce91335d5d1e3df0919ece480","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-10 03:24:52.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"572befca510c23e42061d3d31ed09b624a338d2d","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-11 12:02:08.000000000","message":"Patch Set 9: Code-Review-1\n\n(15 comments)\n\nOverall looks good to me, it would be exciting and easier to understand if a sequence diagram of driver implementation and interaction with Cyborg related components could be added.","accounts_in_message":[],"_revision_number":9},{"id":"1cf303d6ef85f714e5e1ee470e8a90dc0ad46bf2","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-11 12:04:12.000000000","message":"Patch Set 9:\n\n\u003e (15 comments)\n \u003e \n \u003e Overall looks good to me, it would be exciting and easier to\n \u003e understand if a sequence diagram of driver implementation and\n \u003e interaction with Cyborg related components could be added.\n\nAnd it might be coverd by Nova, Placement.","accounts_in_message":[],"_revision_number":9},{"id":"7d5abda876208eefd984a5f35426289294111e68","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-17 09:09:54.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"9b5cde277ba989768411fd7f450341b912ce8d81","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-17 09:12:06.000000000","message":"Patch Set 10:\n\n(17 comments)\n\nThanks, Brin for the review. Please see the update in my new patch.","accounts_in_message":[],"_revision_number":10},{"id":"30883bb21631c6f29c0a3beeb108120ed2e751fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-17 09:18:42.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/091b3b7905f1420fb4838bdcaf2ba45a : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b211c69e7b0a49de8eb5e4019ef3bae7 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":10},{"id":"f21f2aa3b11703fe02f99d710316f65f292ffa66","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-17 09:35:45.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"a614ca315dd7b311b3389b2e08198614453c0c71","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-17 11:34:39.000000000","message":"Patch Set 10: Code-Review+1","accounts_in_message":[],"_revision_number":10},{"id":"233ce47998d5ce30615a826817ad4cbdbd571365","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-19 06:17:05.000000000","message":"Patch Set 10: Code-Review-1\n\n(15 comments)\n\nThanks for your updates. LGTM overall. I think we need to find an agreement with Nova about create and bind mdev first. In my personal opinion, I think it is Cyborg who is responsible for creating mdev because this is in the scope of device management.","accounts_in_message":[],"_revision_number":10},{"id":"a3ec91e140d1ef92bd3769ef9c21c895138f9f77","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-19 06:31:15.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"fa2a8641eee86a7d1d71ac87b0ba1e7bc7a71c99","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-26 09:22:25.000000000","message":"Patch Set 10:\n\n(11 comments)\n\nHi Xinran and Brin, thanks for your review. Please see my update in the new patch.","accounts_in_message":[],"_revision_number":10},{"id":"c7d4b4678fd6c5666fdd7f79efc70f77859aedb6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-26 09:35:45.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"d39d385d1bf2bb5ca4f18af76ae4b204774c024c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-26 09:46:08.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd07bf3d279845fe85a25ec70e4bdbf5 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/634854ae57e041e08876ced9b6b8f9d6 : SUCCESS in 2m 48s","accounts_in_message":[],"_revision_number":11},{"id":"ea0f2b9dbf3b69ce59df0f6865c0b61fa6c6e7b5","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-26 10:00:03.000000000","message":"Patch Set 11: Code-Review-1\n\n(2 comments)\n\nthanks for updating, please see my comments inline.","accounts_in_message":[],"_revision_number":11},{"id":"113e6cbe255fdd29cc3ea41fa44c214c90e8a84c","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-11-27 03:24:19.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nhi xinran, pls check my reply.","accounts_in_message":[],"_revision_number":11},{"id":"303ee091d88be80baefa5a9413a13d76bf2ed1da","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-27 06:13:55.000000000","message":"Patch Set 11:\n\n(1 comment)\n\nlgtm, wait for xinran.","accounts_in_message":[],"_revision_number":11},{"id":"512e2bad32032af81031f77e555cab0313d98dd8","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2020-11-27 08:28:18.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)\n\nlgtm","accounts_in_message":[],"_revision_number":11},{"id":"bc8dc0a1a14671eab5eb31da4ce19a6a7d3b1327","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-27 08:38:28.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"9624bf25b972ee56a0484931b1840dfd83537880","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-27 08:58:43.000000000","message":"Patch Set 11: Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"a9897309333c57579de116118298e3e402068997","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 08:58:56.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"ce26e4bef098cef68198d2c9a67eeeee82f0d881","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 09:07:42.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ae99dec59e94a859ea9f4c28464069b : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef11a468a6d644b0baafe317d0ffcad1 : SUCCESS in 2m 43s","accounts_in_message":[],"_revision_number":11},{"id":"ff4c523449b57764b893f796be7c3f3ba3591386","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 09:07:43.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":11},{"id":"4b32fa031757fea673c9fe0fce2a880f6694fd4c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-27 09:11:07.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/bbe5ac04fca44c55a4aaf4314157b9fe : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"16de8874ccd3c5fa62c7b8e54b3ade7042c66a2b","revisions":{"002e24936b64d9626919931ee9f4daa14e3676e9":{"kind":"REWORK","_number":1,"created":"2020-10-21 03:14:22.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/1"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-21 03:13:59.000000000","tz":480},"subject":"Cyborg NVIDIAGPUDriver support vGPU management","message":"Cyborg NVIDIAGPUDriver support vGPU management\n\nThis spec proposes the enhancement of Cyborg gpu driver to\nsupport vGPU management.\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/002e24936b64d9626919931ee9f4daa14e3676e9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/002e24936b64d9626919931ee9f4daa14e3676e9"}]},"branch":"refs/heads/master"},"dcd08bdb5f43cd3d227d4d6e44136209a38d3ab9":{"kind":"REWORK","_number":2,"created":"2020-10-21 03:39:16.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/2"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-21 03:38:58.000000000","tz":480},"subject":"Cyborg NVIDIAGPUDriver support vGPU management","message":"Cyborg NVIDIAGPUDriver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/dcd08bdb5f43cd3d227d4d6e44136209a38d3ab9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/dcd08bdb5f43cd3d227d4d6e44136209a38d3ab9"}]},"branch":"refs/heads/master"},"004b11d599a5ccf9e3e7d8bc720cfed55308031f":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2020-10-24 03:12:24.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/3"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-24 03:11:17.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/004b11d599a5ccf9e3e7d8bc720cfed55308031f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/004b11d599a5ccf9e3e7d8bc720cfed55308031f"}]},"branch":"refs/heads/master"},"59d6e808e4b5e17688d05659d05216bbd8da0718":{"kind":"REWORK","_number":4,"created":"2020-10-24 07:25:48.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/4"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-24 07:25:11.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/59d6e808e4b5e17688d05659d05216bbd8da0718"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/59d6e808e4b5e17688d05659d05216bbd8da0718"}]},"branch":"refs/heads/master"},"57d372c993b3123d57c3cdcab926413973763dcf":{"kind":"REWORK","_number":5,"created":"2020-11-05 06:36:10.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/5"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-11-05 02:46:08.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/57d372c993b3123d57c3cdcab926413973763dcf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/57d372c993b3123d57c3cdcab926413973763dcf"}]},"branch":"refs/heads/master"},"009dc691d975be9dd76a25115e61243773aff297":{"kind":"REWORK","_number":6,"created":"2020-11-06 12:28:11.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/6"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-11-06 12:21:37.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/009dc691d975be9dd76a25115e61243773aff297"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/009dc691d975be9dd76a25115e61243773aff297"}]},"branch":"refs/heads/master"},"480695f9cc83266292728db05e87a59743476e8c":{"kind":"REWORK","_number":7,"created":"2020-11-06 12:31:15.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/7"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-11-06 12:30:45.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/480695f9cc83266292728db05e87a59743476e8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/480695f9cc83266292728db05e87a59743476e8c"}]},"branch":"refs/heads/master"},"2d8419697b6fc268e52958de70a3b7a85cf7339d":{"kind":"REWORK","_number":8,"created":"2020-11-06 12:33:19.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/8"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-11-06 12:32:53.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2d8419697b6fc268e52958de70a3b7a85cf7339d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/2d8419697b6fc268e52958de70a3b7a85cf7339d"}]},"branch":"refs/heads/master"},"6255f66e4478b2bdf9d2934f796387039dbde168":{"kind":"REWORK","_number":9,"created":"2020-11-09 06:58:39.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/9"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-11-09 06:58:06.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6255f66e4478b2bdf9d2934f796387039dbde168"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6255f66e4478b2bdf9d2934f796387039dbde168"}]},"branch":"refs/heads/master"},"099440721bf8d8c784fe46a675ecddfc732e7846":{"kind":"REWORK","_number":10,"created":"2020-11-17 09:09:54.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/10"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-11-17 09:09:10.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/099440721bf8d8c784fe46a675ecddfc732e7846"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/099440721bf8d8c784fe46a675ecddfc732e7846"}]},"branch":"refs/heads/master"},"16de8874ccd3c5fa62c7b8e54b3ade7042c66a2b":{"kind":"REWORK","_number":11,"created":"2020-11-26 09:35:45.000000000","uploader":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"ref":"refs/changes/25/758925/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/25/758925/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/25/758925/11"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-10-20 09:39:45.000000000","tz":480},"committer":{"name":"Yumeng_Bao","email":"yumeng_bao@yahoo.com","date":"2020-11-26 09:35:35.000000000","tz":480},"subject":"Cyborg NVIDIA GPU Driver support vGPU management","message":"Cyborg NVIDIA GPU Driver support vGPU management\n\nThe cyborg gpu driver has supported pGPU management, this spec proposes\nthe enhancement of Cyborg gpu driver to support vGPU management.\n\nTask: 41126\nStory: 2008265\n\nChange-Id: If44ad617fcf544690dcbaad1183953c68440bc3b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/16de8874ccd3c5fa62c7b8e54b3ade7042c66a2b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/16de8874ccd3c5fa62c7b8e54b3ade7042c66a2b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
