)]}'
{"id":"openstack%2Fcyborg-specs~759557","triplet_id":"openstack%2Fcyborg-specs~master~I9a6941eccebf65da4df90b95a1dc36f9bca40dc1","project":"openstack/cyborg-specs","branch":"master","topic":"759557","hashtags":[],"change_id":"I9a6941eccebf65da4df90b95a1dc36f9bca40dc1","subject":"Add NVIDIA GPU driver specification","status":"MERGED","created":"2020-10-24 09:30:30.000000000","updated":"2021-07-15 12:42:46.000000000","submitted":"2021-07-15 12:41:30.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":18,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"759557-759557","meta_rev_id":"63a963196fc407a47a7f2891f3ac8d9072d5d911","_number":759557,"virtual_id_number":759557,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-15 12:41:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":2,"date":"2021-07-15 12:30:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":2,"date":"2021-07-15 09:26:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"value":1,"date":"2021-07-15 12:30:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-24 09:32:15.000000000","updated_by":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"reviewer":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"state":"REVIEWER"},{"updated":"2020-10-24 09:32:45.000000000","updated_by":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2020-10-24 09:33:30.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2020-11-10 01:13:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-11-11 10:16:23.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"}],"messages":[{"id":"c4204e414e4e558b276b83eab340c897986e0e63","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-10-24 09:30:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4fe41014444904bc5a06dbc5e67fe92ffa42e85d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-24 09:37:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37135355ce134c56858750e959089a96 : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2592e4e61a4df597dd4d4447d6bfb8 : SUCCESS in 2m 54s","accounts_in_message":[],"_revision_number":1},{"id":"28ba62e4ddd5ceb26f51aa0542ddb8c445d548b0","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-10-26 07:27:34.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"453c30ef9abed3577c2d65d42871485f5935bd59","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-04 09:17:37.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"48ad30c225a771fec010063711b0b94512450315","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-04 09:28:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30ee01b62e5144e0b0ec34003c924846 : SUCCESS in 4m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/838f0567e05146698be697d251cb2dd3 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":2},{"id":"ffcaaf6469b7807cd1d1ae85058ebaa1c0ea395b","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2020-11-10 01:00:54.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"383c6ba3fe846a10d130cef31c089fd9efad6ee4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-11-10 01:13:03.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/111732c9f24d4c019f58f54500c103a1 : SUCCESS in 4m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e4967640a794904be9ccb8b9e6f6a02 : SUCCESS in 3m 21s","accounts_in_message":[],"_revision_number":3},{"id":"9b817aedec34b7730ecda4658268c29d756f0f73","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2020-11-11 10:16:23.000000000","message":"Patch Set 3: Code-Review+2\n\nLGTM","accounts_in_message":[],"_revision_number":3},{"id":"7f403a59cdee5f9972ebcd470a235e71f7cd0798","author":{"_account_id":24872,"name":"YumengBao","email":"yumeng_bao@yahoo.com","username":"Yumeng_Bao"},"date":"2020-12-24 02:54:59.000000000","message":"Patch Set 3: Code-Review-1\n\n(5 comments)\n\nThanks wenping for the update. pls check my comments inline.","accounts_in_message":[],"_revision_number":3},{"id":"09bf4e16169266741e849d65bcc09a7fc28f515e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-06-10 07:20:46.000000000","message":"Uploaded patch set 4: New patch set was added with same tree, parent, and commit message as Patch Set 3.","accounts_in_message":[],"_revision_number":4},{"id":"2408fae2e4407e0f2a0740dac73df1a1cfbeb93a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-06-10 07:25:08.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4},{"id":"32b7ae8fb903c59638106a8377a7b4ff741b121e","tag":"autogenerated:gerrit:restore","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-15 06:16:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":4},{"id":"2bc3c1ecb2a66a13c012b7629ac3b9b8f10f1fba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 06:30:20.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/42edfaf2eee748c08f71d258c675319d : SUCCESS in 6m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d1e4c90f7b924059a5cdb940764ed110 : SUCCESS in 6m 18s","accounts_in_message":[],"_revision_number":4},{"id":"c466436ad90de86fe14e4ad2885dc3ee0c1f4f3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-15 06:35:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5c1ffdd069b438cbe798046f925582fd792c3aad","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-15 06:37:41.000000000","message":"Patch Set 5:\n\n(5 comments)","accounts_in_message":[],"_revision_number":5},{"id":"1239e488822e8b6f088fc33e2e5050e41113eb95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 06:45:46.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e2d1835744b4f4fbe5e22c485cd3515 : SUCCESS in 4m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/051c73e3c08d4a4db186de617a1caad8 : SUCCESS in 4m 15s","accounts_in_message":[],"_revision_number":5},{"id":"9867236f5459d43109fedde646be664710e33220","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-07-15 08:11:11.000000000","message":"Patch Set 5: Code-Review+2\n\nLGTM, thanks.","accounts_in_message":[],"_revision_number":5},{"id":"aa590279e4bc32e2234202cc06772c2b1280e626","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-07-15 08:36:36.000000000","message":"Patch Set 5: Code-Review-1\n\n(2 comments)\n\nit looks good to me overall. please check my comments inline. i think it is almost ready to go","accounts_in_message":[],"_revision_number":5},{"id":"747ecc7c260f2f84b74aa20eb7029c0bcf027abd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-15 09:06:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"d0ddf76a7ae3664760e0360667b8cdb6eab6fe2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 09:13:00.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3297275b096d4671afb51270de85c116 : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7a27bec504f24380a31b2004ca7624d8 : SUCCESS in 4m 24s","accounts_in_message":[],"_revision_number":6},{"id":"d5ea6f5663c70b99bbff7ab998407524491fd4ad","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-15 09:14:54.000000000","message":"Patch Set 6:\n\n(2 comments)","accounts_in_message":[],"_revision_number":6},{"id":"91bb1a415d2c49f059b744dd0cb25c34c0f24d12","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-07-15 09:26:05.000000000","message":"Patch Set 6: Code-Review+2\n\nlgtm","accounts_in_message":[],"_revision_number":6},{"id":"e0b008ef18fc7cddf9d0cd4139dd5b6cc304d21e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-07-15 12:30:33.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"5e3f508e189c7c9397feb726d2e564b4e25ada86","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 12:31:02.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3174a6561e0724d8f0c4995c1f49dc79912f2117","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 12:41:29.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03d4b8da85642d1920b12101bfd8cec : SUCCESS in 4m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c48116649be44e281fa59060c1b0c37 : SUCCESS in 4m 14s","accounts_in_message":[],"_revision_number":6},{"id":"487f7e871b9f57da9b2a483bc5e71c2139ad5759","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 12:41:30.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"63a963196fc407a47a7f2891f3ac8d9072d5d911","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-15 12:42:46.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/c5b4623f24614cf292238013510b8b55 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"4281ff28da1cf869b08b417616533b0947a317c1","revisions":{"4ec04924c50a36cd214b9b097b710e4a5d15e41e":{"kind":"REWORK","_number":1,"created":"2020-10-24 09:30:30.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/57/759557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/57/759557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/1"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-24 09:29:28.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-24 09:29:46.000000000","tz":480},"subject":"Add NVIDIA GPU driver specification","message":"Add NVIDIA GPU driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s NVIDIA GPU driver.\n\nChange-Id: I9a6941eccebf65da4df90b95a1dc36f9bca40dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/4ec04924c50a36cd214b9b097b710e4a5d15e41e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/4ec04924c50a36cd214b9b097b710e4a5d15e41e"}]},"branch":"refs/heads/master"},"40ce9f9fee6bb9b1704efede0aa4493c72836f74":{"kind":"REWORK","_number":2,"created":"2020-11-04 09:17:37.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/57/759557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/57/759557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/2"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-24 09:29:28.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-11-04 09:17:20.000000000","tz":480},"subject":"Add NVIDIA GPU driver specification","message":"Add NVIDIA GPU driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s NVIDIA GPU driver.\n\nChange-Id: I9a6941eccebf65da4df90b95a1dc36f9bca40dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/40ce9f9fee6bb9b1704efede0aa4493c72836f74"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/40ce9f9fee6bb9b1704efede0aa4493c72836f74"}]},"branch":"refs/heads/master"},"28d6b88c64d7ff37601255ff7b6f952f71bded78":{"kind":"REWORK","_number":3,"created":"2020-11-10 01:00:54.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/57/759557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/57/759557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/3"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-24 09:29:28.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-11-10 00:59:29.000000000","tz":480},"subject":"Add NVIDIA GPU driver specification","message":"Add NVIDIA GPU driver specification\n\nThis spec proposes to provide the t initial design for\nCyborg\u0027s NVIDIA GPU driver.\n\nChange-Id: I9a6941eccebf65da4df90b95a1dc36f9bca40dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/28d6b88c64d7ff37601255ff7b6f952f71bded78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/28d6b88c64d7ff37601255ff7b6f952f71bded78"}]},"branch":"refs/heads/master"},"0f9cb716718718a2e02e3b8bbc01e0d15662795f":{"kind":"NO_CHANGE","_number":4,"created":"2021-06-10 07:20:46.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/57/759557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/57/759557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/4"}}},"commit":{"parents":[{"commit":"1b52f6defbf88532f2a0bd31d6aa2f8330675cfb","subject":"Move completed victoria features to implemented directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/1b52f6defbf88532f2a0bd31d6aa2f8330675cfb"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-24 09:29:28.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-06-10 07:20:30.000000000","tz":0},"subject":"Add NVIDIA GPU driver specification","message":"Add NVIDIA GPU driver specification\n\nThis spec proposes to provide the t initial design for\nCyborg\u0027s NVIDIA GPU driver.\n\nChange-Id: I9a6941eccebf65da4df90b95a1dc36f9bca40dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0f9cb716718718a2e02e3b8bbc01e0d15662795f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0f9cb716718718a2e02e3b8bbc01e0d15662795f"}]},"branch":"refs/heads/master"},"5bffc4bee7863792e22c48cfb3508244999140e0":{"kind":"REWORK","_number":5,"created":"2021-07-15 06:35:14.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/57/759557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/57/759557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/5"}}},"commit":{"parents":[{"commit":"58e6de783b782d2b4a4ff379fdc7806ee10d4521","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/58e6de783b782d2b4a4ff379fdc7806ee10d4521"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-24 09:29:28.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-07-15 06:34:30.000000000","tz":480},"subject":"Add NVIDIA GPU driver specification","message":"Add NVIDIA GPU driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s NVIDIA pGPU driver.\n\nChange-Id: I9a6941eccebf65da4df90b95a1dc36f9bca40dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5bffc4bee7863792e22c48cfb3508244999140e0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/5bffc4bee7863792e22c48cfb3508244999140e0"}]},"branch":"refs/heads/master"},"4281ff28da1cf869b08b417616533b0947a317c1":{"kind":"REWORK","_number":6,"created":"2021-07-15 09:06:19.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/57/759557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/57/759557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/57/759557/6"}}},"commit":{"parents":[{"commit":"58e6de783b782d2b4a4ff379fdc7806ee10d4521","subject":"Merge \"[ussuri][goal] Update contributor documentation\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/58e6de783b782d2b4a4ff379fdc7806ee10d4521"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2020-10-24 09:29:28.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-07-15 09:06:02.000000000","tz":480},"subject":"Add NVIDIA GPU driver specification","message":"Add NVIDIA GPU driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s NVIDIA pGPU driver.\n\nChange-Id: I9a6941eccebf65da4df90b95a1dc36f9bca40dc1\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/4281ff28da1cf869b08b417616533b0947a317c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/4281ff28da1cf869b08b417616533b0947a317c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
