)]}'
{"id":"openstack%2Fcyborg-specs~789506","triplet_id":"openstack%2Fcyborg-specs~master~I1904efd872d7ab41ce6d835ce8dd8ab199b6a081","project":"openstack/cyborg-specs","branch":"master","topic":"bp/cyborg-rootwrap-daemon-mode","hashtags":[],"change_id":"I1904efd872d7ab41ce6d835ce8dd8ab199b6a081","subject":"Running Cyborg with rootwrap as a daemon","status":"ABANDONED","created":"2021-05-04 07:50:51.000000000","updated":"2021-07-08 01:56:24.000000000","total_comment_count":22,"unresolved_comment_count":4,"has_review_started":true,"meta_rev_id":"5bc3c3ac33f0f8163d0a1e979a5e58f9d950b705","_number":789506,"virtual_id_number":789506,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:check","value":1,"date":"2021-05-17 08:54:51.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-04 07:59:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-13 02:32:34.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"},{"updated":"2021-05-13 02:32:34.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-05-13 02:32:34.000000000","updated_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"}],"messages":[{"id":"41bbcb0e3144822f07def6cc192484085352edbe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-04 07:50:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cc65abfd0ae8ba2d822335c55cdd31af8bec42a6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-04 07:59:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9cea9d637cd2450fb2da1d8ff0327c85 : FAILURE in 5m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2581aef3e16542cd9d08c73c7384e0a0 : FAILURE in 3m 33s","accounts_in_message":[],"_revision_number":1},{"id":"5870aadef831f1ea3a2df7a8b4593afad3abdc17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-08 06:57:49.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"6b7afb6af0303a780b25554a57e170ca74d987cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-08 07:03:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b088e4036b8420ea8a0fdd43c7e5c24 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/14764bd4edc34710841e60cf8827f30a : SUCCESS in 3m 19s","accounts_in_message":[],"_revision_number":2},{"id":"a8a2491258f1c3cd5f02d6fec07f41ce1a5e3275","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-05-14 01:08:33.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\noverall looks good to me, agree the roorwap support is necessary, some comments inline, pls fix.","accounts_in_message":[],"_revision_number":2},{"id":"456ea65ddeb38bd507022dfacb95df39de852eb8","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-05-14 13:45:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":2},{"id":"c48dbc6a1bdc4b20cbe0b444f0c0923b0b8af528","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-17 08:14:40.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"c4fea43ba1b4c615313acba5483bdea2373c41f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:24:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9c1e5b3a57a4e7ca6915746e8741e23 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9debf441123c40369fbdf2aa8b792eac : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":3},{"id":"a9948c8c9871de674f64cbdfcc67101cf08d615f","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-17 08:45:22.000000000","message":"Patch Set 3:\n\n(11 comments)\n\nThanks brin and xinran. addressed your comments inline.","accounts_in_message":[],"_revision_number":3},{"id":"1fc1c6325717ecb07b6507f731362efc3c06c04c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-05-17 08:48:56.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"874cf28faf028941ed1f208929d0164ac5642852","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-17 08:54:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edb8580674a747f69a5f8644ae86fd1e : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/097b0410799b425e8cee7699c49df722 : SUCCESS in 4m 07s","accounts_in_message":[],"_revision_number":4},{"id":"b60658fda5fc34797452962e5306892a9007a874","author":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"date":"2021-06-09 08:56:09.000000000","message":"Patch Set 4:\n\nwe may need more investigation on oslo.privsep","accounts_in_message":[],"_revision_number":4},{"id":"27171e22927735735c4ab46c3151c751a7f2a4fe","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-06-09 08:57:49.000000000","message":"Patch Set 4:\n\n\u003e Patch Set 4:\n\u003e \n\u003e we may need more investigation on oslo.privsep\n\nyes, maybe we donnot need this anymore.","accounts_in_message":[],"_revision_number":4},{"id":"5bc3c3ac33f0f8163d0a1e979a5e58f9d950b705","tag":"autogenerated:gerrit:abandon","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-07-08 01:56:24.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"8a2aabe27b346159ec328a22546ed5b75d70b42a","revisions":{"ec3297b6d807bf8187482d35627c8401ebbbabd5":{"kind":"REWORK","_number":1,"created":"2021-05-04 07:50:51.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/06/789506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/06/789506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/1"}}},"commit":{"parents":[{"commit":"00055cbc5706217eb3483bca74a6606ae6275dcf","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/00055cbc5706217eb3483bca74a6606ae6275dcf"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-04 07:49:52.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-04 07:49:56.000000000","tz":0},"subject":"Running Nova with rootwrap as a daemon","message":"Running Nova with rootwrap as a daemon\n\nChange-Id: I1904efd872d7ab41ce6d835ce8dd8ab199b6a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ec3297b6d807bf8187482d35627c8401ebbbabd5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ec3297b6d807bf8187482d35627c8401ebbbabd5"}]},"branch":"refs/heads/master"},"3ef20bdfc81450dfce232a10ea2c8ddf2b8182b9":{"kind":"REWORK","_number":2,"created":"2021-05-08 06:57:49.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/06/789506/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/06/789506/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/2"}}},"commit":{"parents":[{"commit":"00055cbc5706217eb3483bca74a6606ae6275dcf","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/00055cbc5706217eb3483bca74a6606ae6275dcf"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-04 07:49:52.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-08 06:57:32.000000000","tz":0},"subject":"Running Cyborg with rootwrap as a daemon","message":"Running Cyborg with rootwrap as a daemon\n\nChange-Id: I1904efd872d7ab41ce6d835ce8dd8ab199b6a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3ef20bdfc81450dfce232a10ea2c8ddf2b8182b9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3ef20bdfc81450dfce232a10ea2c8ddf2b8182b9"}]},"branch":"refs/heads/master"},"78d6574f991d24d55196fe2cd56ba41093e3bd1c":{"kind":"REWORK","_number":3,"created":"2021-05-17 08:14:40.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/06/789506/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/06/789506/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/3"}}},"commit":{"parents":[{"commit":"00055cbc5706217eb3483bca74a6606ae6275dcf","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/00055cbc5706217eb3483bca74a6606ae6275dcf"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-04 07:49:52.000000000","tz":0},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-17 08:14:25.000000000","tz":0},"subject":"Running Cyborg with rootwrap as a daemon","message":"Running Cyborg with rootwrap as a daemon\n\nChange-Id: I1904efd872d7ab41ce6d835ce8dd8ab199b6a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/78d6574f991d24d55196fe2cd56ba41093e3bd1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/78d6574f991d24d55196fe2cd56ba41093e3bd1c"}]},"branch":"refs/heads/master"},"8a2aabe27b346159ec328a22546ed5b75d70b42a":{"kind":"REWORK","_number":4,"created":"2021-05-17 08:48:56.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/06/789506/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/06/789506/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/06/789506/4"}}},"commit":{"parents":[{"commit":"00055cbc5706217eb3483bca74a6606ae6275dcf","subject":"setup.cfg: Replace dashes with underscores","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/00055cbc5706217eb3483bca74a6606ae6275dcf"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2021-05-04 07:49:52.000000000","tz":0},"committer":{"name":"Wenping Song","email":"songwenping@inspur.com","date":"2021-05-17 08:48:52.000000000","tz":0},"subject":"Running Cyborg with rootwrap as a daemon","message":"Running Cyborg with rootwrap as a daemon\n\nThis blueprint proposes adopting functionality in oslo.rootwrap that\nwould allow to run rootwrap daemon.\n\nImplements: blueprint cyborg-rootwrap-daemon-mode\n\nChange-Id: I1904efd872d7ab41ce6d835ce8dd8ab199b6a081\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/8a2aabe27b346159ec328a22546ed5b75d70b42a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/8a2aabe27b346159ec328a22546ed5b75d70b42a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
