)]}'
{"id":"openstack%2Fcyborg-specs~803103","triplet_id":"openstack%2Fcyborg-specs~master~I2e264cce7d059f133c8a2c5eb3e86e55f94bc268","project":"openstack/cyborg-specs","branch":"master","topic":"bp/add-xilinx-fpga-driver","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2021-12-18 08:43:47.000000000","reason":"Wenping Song replied on the change"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2021-11-09 08:35:18.000000000","reason":"Brin Zhang replied on the change"},"23950":{"account":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"last_update":"2021-12-20 05:39:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I2e264cce7d059f133c8a2c5eb3e86e55f94bc268","subject":"Add Xilinx FPGA driver specification","status":"MERGED","created":"2021-07-31 08:19:20.000000000","updated":"2021-12-20 05:41:15.000000000","submitted":"2021-12-20 05:39:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":36,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"803103-bp/add-xilinx-fpga-driver","meta_rev_id":"c38378f65846aa1405131cf82419b24bc0f9a36c","_number":803103,"virtual_id_number":803103,"owner":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-20 05:39:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2021-12-18 08:43:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2021-11-09 08:35:18.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":1,"date":"2021-12-20 05:34:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"}],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-07-31 08:28:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-07-31 09:07:04.000000000","updated_by":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"reviewer":{"_account_id":25738,"name":"Xinran WANG","email":"xin-ran.wang@intel.com","username":"Xinran"},"state":"REVIEWER"},{"updated":"2021-07-31 09:07:04.000000000","updated_by":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2021-07-31 09:07:04.000000000","updated_by":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"},{"updated":"2021-08-11 12:34:08.000000000","updated_by":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"reviewer":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"state":"REVIEWER"}],"messages":[{"id":"596d1a00f6ebb107430a3af8e3d18626fc88b13a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-07-31 08:19:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8d639a308ac28d19c7b1334681ec1dd86b7d9d05","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-31 08:28:33.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ecd8ae790114875ad44ef3e11f09672 : FAILURE in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a728a98010e24d2884829bb9e296354a : SUCCESS in 3m 46s","accounts_in_message":[],"_revision_number":1},{"id":"c2e3f77c37568b061c63558a5853ea0c3e992ed5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-07-31 08:55:24.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"f50c54c18542aa7d6e7d9a379164c3732bd03f2a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-31 09:04:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b6e8ac72a944d2ea4371648ca62954f : SUCCESS in 4m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/869e4366a9634af2994dc44ddfb946e5 : SUCCESS in 7m 25s","accounts_in_message":[],"_revision_number":2},{"id":"263480097ea320d6506dbf988e624805f11c827e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-08-11 08:40:49.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"15473715b871400b794fcea5f4452e697084d63d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-11 08:47:21.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b087edfc95774c79adef0ee1b85bfd3a : SUCCESS in 4m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d64e3a71954347d6b995498719ce3473 : SUCCESS in 3m 54s","accounts_in_message":[],"_revision_number":3},{"id":"78703569788b3601e0cd95554391f2f780bef80c","author":{"_account_id":28748,"name":"chenker","email":"chen.ke14@zte.com.cn","username":"chenke"},"date":"2021-08-11 12:34:08.000000000","message":"Patch Set 3:\n\nlooks good, need other\u0027s review.","accounts_in_message":[],"_revision_number":3},{"id":"832ab5a621844efc120c307e6ff35235bd9a0f7e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-08-13 01:59:06.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"58dab6f40bba03d2fdf68cc89c095be970bef2d6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-13 02:04:49.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f32e3c915e84776acbfa401640cee76 : SUCCESS in 4m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e340681cddf46819ef32269437bea08 : SUCCESS in 3m 24s","accounts_in_message":[],"_revision_number":4},{"id":"3e5ce11e82f6185ffe45db7f33e251f6b237d78d","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-08-13 02:40:16.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"dfc0631e5dc572afba257d41f7a5d76cd6d835d1","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-08-15 07:28:19.000000000","message":"Patch Set 4: Code-Review-1\n\n(4 comments)\n\nLooks good to me, but the feature of Xena version has been frozen. We can fast approve this spec in Yoga release.\n\nA question: Where is the fpga program() function? or why we cannot support it now?","accounts_in_message":[],"_revision_number":4},{"id":"186270ad353b69e89046a32be9504af2dac3290e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-08-17 10:25:12.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"bd776a1d25cf5610b52e9dd2754475725aa3f927","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-08-17 10:26:42.000000000","message":"Patch Set 5:\n\n(4 comments)","accounts_in_message":[],"_revision_number":5},{"id":"176aca60b754befd788a7f3d614a59fed55a1d2f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 10:32:52.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2897a23f313645059a00b8d9510afd3b : FAILURE in 5m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45c90ea6a72c4cf485560ec75b231649 : FAILURE in 3m 23s\n\nWarning:\n  Job openstack-tox-docs: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path doc/source/specs/yoga/approved/add-xilinx-fpga-driver.rst in the commit\u0027\n  Comments left for invalid file doc/source/specs/yoga/approved/add-xilinx-fpga-driver.rst","accounts_in_message":[],"_revision_number":5},{"id":"8cadc2c244905232b23d29c3513d72f2b7126be1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-08-17 12:14:31.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"f567c9b374c39612bbb3205c57d7f4fbfbd910a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-08-17 12:24:41.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba72e4ec3cd4e8e83999e530af6b2f1 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c27448d6d7e487292ea283f0e1232a7 : SUCCESS in 5m 12s","accounts_in_message":[],"_revision_number":6},{"id":"b00fb1ea3e61d21d3edee04f32c76eb3c47a40ef","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-10-20 01:42:11.000000000","message":"Patch Set 6: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":6},{"id":"03063d20f3b3814122ab956e4587ed61fc133ef6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-10-20 08:50:18.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"7bb4e90541366623130ce2ec8e2508f83340e452","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-10-28 08:19:36.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"05b3bd3af6abe680a4148830767220f0d235806c","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-10-28 08:20:30.000000000","message":"Patch Set 7:\n\n(5 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6f63ce3255294f23b685c52c5b4a02f597a6783d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:28:00.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6ef779e2822480dbde5b859c602f76d : FAILURE in 6m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8f66f52a3b491e9cb5e6683b0a5f10 : FAILURE in 5m 13s","accounts_in_message":[],"_revision_number":7},{"id":"079061f8ef8dc0044858a07697e3eac10bc9273d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-10-28 08:30:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"4d27e20922e137547eb7a3c2a107520503cadce4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-28 08:42:40.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67d6d54f7c2f4f0f988c66a027b2c9db : FAILURE in 4m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67547db4bfc34bfc878573c5ef3099bb : FAILURE in 4m 14s","accounts_in_message":[],"_revision_number":8},{"id":"40de6ee77141297af1697f5e1905469de732485c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-10-29 07:10:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"36030bd0154a0e9b79bb3c154d8c326a777ab522","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 07:20:11.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0a61d5908069446397469d6f76622f2a : SUCCESS in 4m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdbfdb524ac348aebe3d7e018fb2a819 : SUCCESS in 6m 27s","accounts_in_message":[],"_revision_number":9},{"id":"58bf8dd8c023aea73e42866344b5cfe3b2ffc635","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-10-29 07:52:17.000000000","message":"Patch Set 9: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"d84ec68431090b7d422d139210ddcf19ddd552b4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-11-01 07:28:31.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"5483de4e5d06122c726435403c7dac434efc4310","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-11-01 07:32:32.000000000","message":"Patch Set 10:\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"51fd4c3e4c0fe78075e1225b12a7959869d7725d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-01 07:33:52.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1afe6a4ad7554c27a360bad5b0fcf40f : SUCCESS in 4m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0eb153abfdac4b208481ef98a5bce930 : SUCCESS in 3m 51s","accounts_in_message":[],"_revision_number":10},{"id":"fa433d9865e6016d0756512ce1fbb6abe11cef42","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-11-03 08:50:33.000000000","message":"Topic change-803103 removed","accounts_in_message":[],"_revision_number":10},{"id":"534f241ffbd0a3ae39a8b8734b2ce240654f7e83","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-11-03 08:50:41.000000000","message":"Topic set to bp/add-xilinx-fpga-driver","accounts_in_message":[],"_revision_number":10},{"id":"4b5b115958909ac3780a8b42542b4a7602437160","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-11-09 08:35:18.000000000","message":"Patch Set 10: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f19d9645e60e90a06c63ce46da4f04bd09af6dc6","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-11-09 09:05:57.000000000","message":"Patch Set 10: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"8c6c5e17133c52bf3176ebc4fb270adf305b3bd7","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-12-07 10:01:06.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"36c9fb5bbff12e0141949ea72ff182ec1e6615b9","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-12-17 07:42:57.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"bdf49eee4dc08a43a02a684ceb9945a9272cf663","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2021-12-18 08:30:59.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"03c070e6c5da8b7bb070db7d37ff8d8686c80884","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2021-12-18 08:43:47.000000000","message":"Patch Set 10: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"77de1fd51aced7ba105ab5228170656ed49f78de","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2021-12-20 05:34:05.000000000","message":"Patch Set 10: Workflow+1","accounts_in_message":[],"_revision_number":10},{"id":"b72d39d2c8bcfd0d093373267e2bf050ca4f885c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 05:34:37.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"f5e4a4fb72267b496569b435a0636c41aebe11e7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 05:39:35.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d9ef830d80684abead71ffe29fda798f : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b140adb8d454aa6b05749c7deda9fe9 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":10},{"id":"f6747f99755b7072bf587d822cbc1c9520fef492","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 05:39:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":10},{"id":"c38378f65846aa1405131cf82419b24bc0f9a36c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-20 05:41:15.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/d2b4efe04100428bbea2b5f78d97ed37 : SUCCESS in 56s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"3daa899968cb6da69723c0005befb1a01cd8394c","revisions":{"a2a108181fd39746e5a30326fc419f07c60f6723":{"kind":"REWORK","_number":1,"created":"2021-07-31 08:19:20.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/1"}}},"commit":{"parents":[{"commit":"389941434445ead50fae71c21f3aa7d939a90212","subject":"Merge \"Add disable/enable device API.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/389941434445ead50fae71c21f3aa7d939a90212"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a2a108181fd39746e5a30326fc419f07c60f6723"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a2a108181fd39746e5a30326fc419f07c60f6723"}]},"branch":"refs/heads/master"},"6fb5a3e26199a5a35282bfe19620b53ecb4bb84b":{"kind":"REWORK","_number":2,"created":"2021-07-31 08:55:24.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/2"}}},"commit":{"parents":[{"commit":"389941434445ead50fae71c21f3aa7d939a90212","subject":"Merge \"Add disable/enable device API.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/389941434445ead50fae71c21f3aa7d939a90212"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:55:17.000000000","tz":0},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6fb5a3e26199a5a35282bfe19620b53ecb4bb84b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6fb5a3e26199a5a35282bfe19620b53ecb4bb84b"}]},"branch":"refs/heads/master"},"0b85c27a93f98bcc161ad6fdb271e73d6124544e":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2021-08-11 08:40:49.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/3"}}},"commit":{"parents":[{"commit":"389941434445ead50fae71c21f3aa7d939a90212","subject":"Merge \"Add disable/enable device API.\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/389941434445ead50fae71c21f3aa7d939a90212"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-08-11 08:39:21.000000000","tz":0},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0b85c27a93f98bcc161ad6fdb271e73d6124544e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0b85c27a93f98bcc161ad6fdb271e73d6124544e"}]},"branch":"refs/heads/master"},"6fa4f6fdf9c2073f39d3c11ae8e1907c973f3e8a":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2021-08-13 01:59:06.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/4"}}},"commit":{"parents":[{"commit":"3540e96f62f14550c4950cf832bc4ff61be84a45","subject":"Create specs directory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3540e96f62f14550c4950cf832bc4ff61be84a45"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"Eric Xie","email":"eric_xiett@163.com","date":"2021-08-13 01:59:06.000000000","tz":0},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6fa4f6fdf9c2073f39d3c11ae8e1907c973f3e8a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6fa4f6fdf9c2073f39d3c11ae8e1907c973f3e8a"}]},"branch":"refs/heads/master","description":"Rebase"},"767b8154f19fa7f46726934346ff0fd17babe11c":{"kind":"REWORK","_number":5,"created":"2021-08-17 10:25:12.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/5"}}},"commit":{"parents":[{"commit":"3540e96f62f14550c4950cf832bc4ff61be84a45","subject":"Create specs directory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3540e96f62f14550c4950cf832bc4ff61be84a45"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-08-17 10:24:21.000000000","tz":480},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\nSigned-off-by: ericxiett \u003ceric_xiett@163.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/767b8154f19fa7f46726934346ff0fd17babe11c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/767b8154f19fa7f46726934346ff0fd17babe11c"}]},"branch":"refs/heads/master"},"78bda6ae84e064d5bf77adc661bcbdc27cf069d5":{"kind":"REWORK","_number":6,"created":"2021-08-17 12:14:31.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/6"}}},"commit":{"parents":[{"commit":"3540e96f62f14550c4950cf832bc4ff61be84a45","subject":"Create specs directory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3540e96f62f14550c4950cf832bc4ff61be84a45"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-08-17 12:13:55.000000000","tz":480},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\nSigned-off-by: ericxiett \u003ceric_xiett@163.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/78bda6ae84e064d5bf77adc661bcbdc27cf069d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/78bda6ae84e064d5bf77adc661bcbdc27cf069d5"}]},"branch":"refs/heads/master"},"0e014afb9f3ff9bb77725d0eea05a9a2fb11a6ec":{"kind":"REWORK","_number":7,"created":"2021-10-28 08:19:36.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/7"}}},"commit":{"parents":[{"commit":"3540e96f62f14550c4950cf832bc4ff61be84a45","subject":"Create specs directory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3540e96f62f14550c4950cf832bc4ff61be84a45"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-10-28 08:18:26.000000000","tz":480},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\nSigned-off-by: ericxiett \u003ceric_xiett@163.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0e014afb9f3ff9bb77725d0eea05a9a2fb11a6ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0e014afb9f3ff9bb77725d0eea05a9a2fb11a6ec"}]},"branch":"refs/heads/master"},"b8aa51ff647e494d9c4589be3e3320350f3d0d3c":{"kind":"REWORK","_number":8,"created":"2021-10-28 08:30:06.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/8"}}},"commit":{"parents":[{"commit":"3540e96f62f14550c4950cf832bc4ff61be84a45","subject":"Create specs directory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3540e96f62f14550c4950cf832bc4ff61be84a45"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-10-28 08:28:51.000000000","tz":480},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\nSigned-off-by: ericxiett \u003ceric_xiett@163.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b8aa51ff647e494d9c4589be3e3320350f3d0d3c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b8aa51ff647e494d9c4589be3e3320350f3d0d3c"}]},"branch":"refs/heads/master"},"cf307ea1bec16bafe274fa77967174ceceed924f":{"kind":"REWORK","_number":9,"created":"2021-10-29 07:10:36.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/9"}}},"commit":{"parents":[{"commit":"3540e96f62f14550c4950cf832bc4ff61be84a45","subject":"Create specs directory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3540e96f62f14550c4950cf832bc4ff61be84a45"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-10-29 07:09:23.000000000","tz":480},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\nSigned-off-by: ericxiett \u003ceric_xiett@163.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/cf307ea1bec16bafe274fa77967174ceceed924f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/cf307ea1bec16bafe274fa77967174ceceed924f"}]},"branch":"refs/heads/master"},"3daa899968cb6da69723c0005befb1a01cd8394c":{"kind":"REWORK","_number":10,"created":"2021-11-01 07:28:31.000000000","uploader":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"ref":"refs/changes/03/803103/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/03/803103/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/03/803103/10"}}},"commit":{"parents":[{"commit":"3540e96f62f14550c4950cf832bc4ff61be84a45","subject":"Create specs directory for Yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3540e96f62f14550c4950cf832bc4ff61be84a45"}]}],"author":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-07-31 08:18:13.000000000","tz":0},"committer":{"name":"ericxiett","email":"eric_xiett@163.com","date":"2021-11-01 07:28:29.000000000","tz":0},"subject":"Add Xilinx FPGA driver specification","message":"Add Xilinx FPGA driver specification\n\nThis spec proposes to provide the initial design for\nCyborg\u0027s Xilinx FPGA driver.\n\nImplements: blueprint add-xilinx-fpga-driver\nChange-Id: I2e264cce7d059f133c8a2c5eb3e86e55f94bc268\nSigned-off-by: ericxiett \u003ceric_xiett@163.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3daa899968cb6da69723c0005befb1a01cd8394c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/3daa899968cb6da69723c0005befb1a01cd8394c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
