)]}'
{"id":"openstack%2Fcyborg-specs~836898","triplet_id":"openstack%2Fcyborg-specs~master~I08c8abf43db2940b256fd52f074f53daf9c6847d","project":"openstack/cyborg-specs","branch":"master","topic":"create_zed_dirs","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2022-04-08 01:05:03.000000000","reason":"Wenping Song replied on the change"},"26458":{"account":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"last_update":"2022-04-08 01:23:15.000000000","reason":"Change was submitted"},"23950":{"account":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"last_update":"2022-04-07 03:40:43.000000000","reason":"Eric Xie replied on the change"}},"hashtags":[],"change_id":"I08c8abf43db2940b256fd52f074f53daf9c6847d","subject":"Create specs directory for Zed","status":"MERGED","created":"2022-04-07 03:19:39.000000000","updated":"2022-04-08 01:24:29.000000000","submitted":"2022-04-08 01:23:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"836898-create_zed_dirs","meta_rev_id":"d240cc9bc28fcf266c4b314bdca999e70c5f0019","_number":836898,"virtual_id_number":836898,"owner":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-04-08 01:23:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":2,"date":"2022-04-08 01:05:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-04-07 03:40:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"all":[{"value":1,"date":"2022-04-08 01:05:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-04-07 03:24:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-04-07 03:30:37.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"state":"REVIEWER"},{"updated":"2022-04-07 03:30:37.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"state":"REVIEWER"}],"messages":[{"id":"a8f6bd238b381a0346098a1a182e74f98fd8489e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-04-07 03:19:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7c019bd984d179761d622f31b5a280babfd40dbc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 03:24:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4ad6cac06e24ddc92ca33cc5b21ba4c : FAILURE in 3m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/294d4afc78324175be4548b45e1f8e0f : FAILURE in 3m 15s","accounts_in_message":[],"_revision_number":1},{"id":"a729e5a7b390c61827824cc5a5c408f06b38af85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-04-07 03:24:56.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b2e3c36f2f325bbbc60e3db6768d6e4c466ca61a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2022-04-07 03:26:45.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"ee705e51900d577457e106ef375d49d9ed335f2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-07 03:32:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0ab9f646bc64f01b42d3655318369bb : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89df873db0dd430fb34ebc36cbf26e71 : SUCCESS in 3m 57s","accounts_in_message":[],"_revision_number":3},{"id":"412345c9fa11f2d7aa19eb94e4035ddaa9f3a01b","author":{"_account_id":23950,"name":"Eric Xie","email":"eric_xiett@163.com","username":"ericxie"},"date":"2022-04-07 03:40:43.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"34b87ad8af85864beac01363b83a51e4678d9002","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2022-04-08 01:05:03.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"dc26f92950667cd0171990377e0443e0acb80546","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 01:05:31.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"37d89e0ce0a13a9e23a0c3717709b9efe594aa71","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 01:23:14.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa347de3bae24cf4a353e8102d67444c : SUCCESS in 5m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ea4049827fe485fb4f64241eff957e8 : SUCCESS in 3m 30s","accounts_in_message":[],"_revision_number":3},{"id":"417176aa7fa41a391165660ed48d965f0c3d7daf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 01:23:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"d240cc9bc28fcf266c4b314bdca999e70c5f0019","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-04-08 01:24:29.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/ba0cca42ab614fe7a0166a8abe97ba73 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ed02c8918691c790325447a97a1c06db61c345be","revisions":{"67d0dd93f7b4912545f00a6e25fe3f853f3ca1e2":{"kind":"REWORK","_number":1,"created":"2022-04-07 03:19:39.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/98/836898/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/98/836898/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/1"}}},"commit":{"parents":[{"commit":"77cf073d5450aa00fe54312befa5d6d108cb3c01","subject":"Fix the spec docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/77cf073d5450aa00fe54312befa5d6d108cb3c01"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-04-07 03:19:03.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-04-07 03:19:38.000000000","tz":480},"subject":"Create specs directory for Zed","message":"Create specs directory for Zed\n\nThis patch creates new dirs for Zed and new templates.\n\nChange-Id: I08c8abf43db2940b256fd52f074f53daf9c6847d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/67d0dd93f7b4912545f00a6e25fe3f853f3ca1e2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/67d0dd93f7b4912545f00a6e25fe3f853f3ca1e2"}]},"branch":"refs/heads/master"},"e0d3f8aa3f8d53e778f25da111007f4ff3dcd6de":{"kind":"REWORK","_number":2,"created":"2022-04-07 03:24:56.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/98/836898/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/98/836898/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/2"}}},"commit":{"parents":[{"commit":"77cf073d5450aa00fe54312befa5d6d108cb3c01","subject":"Fix the spec docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/77cf073d5450aa00fe54312befa5d6d108cb3c01"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-04-07 03:19:03.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-04-07 03:24:17.000000000","tz":480},"subject":"Create specs directory for Zed","message":"Create specs directory for Zed\n\nThis patch creates new dirs for Zed and new templates.\n\nChange-Id: I08c8abf43db2940b256fd52f074f53daf9c6847d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e0d3f8aa3f8d53e778f25da111007f4ff3dcd6de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/e0d3f8aa3f8d53e778f25da111007f4ff3dcd6de"}]},"branch":"refs/heads/master"},"ed02c8918691c790325447a97a1c06db61c345be":{"kind":"REWORK","_number":3,"created":"2022-04-07 03:26:45.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/98/836898/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/98/836898/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/98/836898/3"}}},"commit":{"parents":[{"commit":"77cf073d5450aa00fe54312befa5d6d108cb3c01","subject":"Fix the spec docs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/77cf073d5450aa00fe54312befa5d6d108cb3c01"}]}],"author":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-04-07 03:19:03.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2022-04-07 03:26:39.000000000","tz":480},"subject":"Create specs directory for Zed","message":"Create specs directory for Zed\n\nThis patch creates new dirs for Zed and new templates.\n\nChange-Id: I08c8abf43db2940b256fd52f074f53daf9c6847d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ed02c8918691c790325447a97a1c06db61c345be"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ed02c8918691c790325447a97a1c06db61c345be"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
