)]}'
{"id":"openstack%2Fcyborg-specs~877153","triplet_id":"openstack%2Fcyborg-specs~master~Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10","project":"openstack/cyborg-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2023-03-18 09:54:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10","subject":"2023.2.Bobcat spec maintain","status":"MERGED","created":"2023-03-11 08:54:44.000000000","updated":"2023-03-18 09:54:55.000000000","submitted":"2023-03-18 09:54:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877153","meta_rev_id":"e4f431ef81f8303568da973a7dace26a40e01b85","_number":877153,"virtual_id_number":877153,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-18 09:54:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":2,"date":"2023-03-18 09:49:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"all":[{"value":1,"date":"2023-03-18 09:49:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-11 09:00:22.000000000","updated_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"reviewer":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"state":"REVIEWER"},{"updated":"2023-03-11 09:18:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"25e16d86bfdc3493536085227e2e8a346d37c64d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-03-11 08:54:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51430d5c5f16d4789f35f5d5f3374fce9443d8a6","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2023-03-11 09:00:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"da050712fc3f0bd9913f1e783437dd17aca194c2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-11 09:18:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d01d0ddbc8ee4f3b9e84e6ae1e1b85b1\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/afebc5ceed554aeb9db077dde979453a : FAILURE in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/862961e719db419caf8ff4640da9039e : FAILURE in 2m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6c5fcb00c0a2e5b4e2dcd48753698e7c6ac2284c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-03-16 10:03:16.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"08f3e94267f0139314d5b3da9557c6c3c2944992","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 10:12:17.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b1e0f42c958429d9e9e10c03b610eae\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6244ea6c278049bdae454070016b03f5 : FAILURE in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18342aebd00c4f99bf8d85edc8b247ab : SUCCESS in 2m 49s","accounts_in_message":[],"_revision_number":2},{"id":"82070b6395e4cbd371599c19c9a7fc8fecc77b10","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-03-17 09:42:03.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"6bdecf9e3e33120ea79679ae3acd33b084fb3be2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 09:47:05.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f5686cfdab4040c188887594cd5a111f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e685bd24551f45d9b0af700a315bc4fc : FAILURE in 2m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/699ea985bd6b403d820f53e9bc0151c6 : SUCCESS in 3m 00s","accounts_in_message":[],"_revision_number":3},{"id":"8d9037f30e0c915f4511b636bb7928695ada9d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-03-17 10:04:16.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"225698884b11699d2aa6c0f9ff0e2871bb1a7042","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 10:16:07.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3f46b2b5375242b2868670011a30c0a9\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8841c47efcb9454fbb5bea7fd0cf82bc : FAILURE in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b1d490361b546bdb8344da87118c8d5 : SUCCESS in 3m 33s","accounts_in_message":[],"_revision_number":4},{"id":"1e5eb4f471bad3a4d7d18666f114b893c9622691","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-03-17 10:23:32.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"a7515cad67ba0062b7769fa8347f3afc2723ffc9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-17 10:33:19.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e744071ca9b4726b691a88adf8a6592\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e455b9f60aca47c3a927d294dd820fc2 : FAILURE in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad07672c03f04ea5b4a27ac2839cf07e : FAILURE in 2m 47s","accounts_in_message":[],"_revision_number":5},{"id":"85100562b3e9347059a3f22e5a7e9beb649f7241","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2023-03-18 09:34:06.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"236e0330b100448d9cab400759c2f37e1f5bcfef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 09:39:09.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5cd5207ac20490dbac5541a2c06a880\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a915517b3bf244bda36824aa3615a286 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d998ab56ce249d282f9399c20335a46 : SUCCESS in 2m 14s","accounts_in_message":[],"_revision_number":6},{"id":"a2f7b8ae88eb4520952682dd9394e49b40e6f81e","author":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"date":"2023-03-18 09:49:55.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"71a4ba30e1552d1bd3576fb0ed6369d08383cc7c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 09:50:14.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"451099c155cb6d879d337e5b2eee7b6bb279ed1b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 09:54:02.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b84e5efdeea49839a8601c613dc5975\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43f383496aed41398879ba2db490c295 : SUCCESS in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c5d34f46fc9409fa6b1f42c67b3412c : SUCCESS in 2m 09s","accounts_in_message":[],"_revision_number":6},{"id":"c40b0a1e067c4f6b9709e59155565b705fd0fa49","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 09:54:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"e4f431ef81f8303568da973a7dace26a40e01b85","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-18 09:54:55.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/379aaffb6e9c44b48758a250e1cc53a5\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/555a5092ca804487adf6522831548ca7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"253be5b31a80097fccac2a4579ffa35b27f3d8d5","revisions":{"fd77a7961043b83711d9404e11f6fcbe676c8a5d":{"kind":"REWORK","_number":1,"created":"2023-03-11 08:54:44.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/53/877153/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/53/877153/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/1"}}},"commit":{"parents":[{"commit":"eb7650805057fc045a3ff8c42c1915f0da5e3db1","subject":"Maintain specs of approved and completed of Antelop version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/eb7650805057fc045a3ff8c42c1915f0da5e3db1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-11 08:54:35.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-11 08:54:35.000000000","tz":480},"subject":"2023.2.Bobcat spec maintain","message":"2023.2.Bobcat spec maintain\n\nChange-Id: Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/fd77a7961043b83711d9404e11f6fcbe676c8a5d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/fd77a7961043b83711d9404e11f6fcbe676c8a5d"}]},"branch":"refs/heads/master"},"ad740818a974270181232041bc1943852cd4160e":{"kind":"REWORK","_number":2,"created":"2023-03-16 10:03:16.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/53/877153/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/53/877153/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/2"}}},"commit":{"parents":[{"commit":"eb7650805057fc045a3ff8c42c1915f0da5e3db1","subject":"Maintain specs of approved and completed of Antelop version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/eb7650805057fc045a3ff8c42c1915f0da5e3db1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-11 08:54:35.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-16 10:02:35.000000000","tz":480},"subject":"2023.2.Bobcat spec maintain","message":"2023.2.Bobcat spec maintain\n\nChange-Id: Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ad740818a974270181232041bc1943852cd4160e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/ad740818a974270181232041bc1943852cd4160e"}]},"branch":"refs/heads/master"},"7376ce39f8d80cd3f21e24e9b3009c25457e7f4d":{"kind":"REWORK","_number":3,"created":"2023-03-17 09:42:03.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/53/877153/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/53/877153/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/3"}}},"commit":{"parents":[{"commit":"eb7650805057fc045a3ff8c42c1915f0da5e3db1","subject":"Maintain specs of approved and completed of Antelop version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/eb7650805057fc045a3ff8c42c1915f0da5e3db1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-11 08:54:35.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-17 09:41:46.000000000","tz":480},"subject":"2023.2.Bobcat spec maintain","message":"2023.2.Bobcat spec maintain\n\nChange-Id: Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/7376ce39f8d80cd3f21e24e9b3009c25457e7f4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/7376ce39f8d80cd3f21e24e9b3009c25457e7f4d"}]},"branch":"refs/heads/master"},"0a6a6783d00ad3cebbb6a31b63c7423d06681b90":{"kind":"REWORK","_number":4,"created":"2023-03-17 10:04:16.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/53/877153/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/53/877153/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/4"}}},"commit":{"parents":[{"commit":"eb7650805057fc045a3ff8c42c1915f0da5e3db1","subject":"Maintain specs of approved and completed of Antelop version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/eb7650805057fc045a3ff8c42c1915f0da5e3db1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-11 08:54:35.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-17 10:04:06.000000000","tz":480},"subject":"2023.2.Bobcat spec maintain","message":"2023.2.Bobcat spec maintain\n\nChange-Id: Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a6a6783d00ad3cebbb6a31b63c7423d06681b90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/0a6a6783d00ad3cebbb6a31b63c7423d06681b90"}]},"branch":"refs/heads/master"},"d885b729869c45a62d7e385835432dfe0352bc87":{"kind":"REWORK","_number":5,"created":"2023-03-17 10:23:32.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/53/877153/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/53/877153/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/5"}}},"commit":{"parents":[{"commit":"eb7650805057fc045a3ff8c42c1915f0da5e3db1","subject":"Maintain specs of approved and completed of Antelop version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/eb7650805057fc045a3ff8c42c1915f0da5e3db1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-11 08:54:35.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-17 10:23:18.000000000","tz":480},"subject":"2023.2.Bobcat spec maintain","message":"2023.2.Bobcat spec maintain\n\nChange-Id: Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/d885b729869c45a62d7e385835432dfe0352bc87"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/d885b729869c45a62d7e385835432dfe0352bc87"}]},"branch":"refs/heads/master"},"253be5b31a80097fccac2a4579ffa35b27f3d8d5":{"kind":"REWORK","_number":6,"created":"2023-03-18 09:34:06.000000000","uploader":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"},"ref":"refs/changes/53/877153/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/53/877153/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/53/877153/6"}}},"commit":{"parents":[{"commit":"eb7650805057fc045a3ff8c42c1915f0da5e3db1","subject":"Maintain specs of approved and completed of Antelop version","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/eb7650805057fc045a3ff8c42c1915f0da5e3db1"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-03-11 08:54:35.000000000","tz":480},"committer":{"name":"zhangbailin","email":"zhangbailin@inspur.com","date":"2023-03-18 09:34:04.000000000","tz":0},"subject":"2023.2.Bobcat spec maintain","message":"2023.2.Bobcat spec maintain\n\nChange-Id: Ia8c9b0591fdd4b5eb35f65d9aae0fff46b820f10\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/253be5b31a80097fccac2a4579ffa35b27f3d8d5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/253be5b31a80097fccac2a4579ffa35b27f3d8d5"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":26458,"name":"Brin Zhang","email":"zhangbailin@inspur.com","username":"zhangbailin"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
