)]}'
{"id":"openstack%2Fcyborg-specs~896076","triplet_id":"openstack%2Fcyborg-specs~master~Iea588f6b9499c35e9a60f4a2012cf6443b5e7586","project":"openstack/cyborg-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"31412":{"account":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"last_update":"2026-06-23 14:22:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iea588f6b9499c35e9a60f4a2012cf6443b5e7586","subject":"move attribute spec to impletment","status":"MERGED","created":"2023-09-26 11:49:51.000000000","updated":"2026-06-23 14:23:34.000000000","submitted":"2026-06-23 14:22:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896076","meta_rev_id":"1c9bc18acd0978545a4ea982dc5d884ad95062d8","_number":896076,"virtual_id_number":896076,"owner":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-06-23 14:22:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-06-23 13:57:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-06-23 13:57:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-26 11:59:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-06-23 13:57:07.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"}],"messages":[{"id":"127a84b9a28f7137adf7020529115258fa818ac8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"date":"2023-09-26 11:49:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"58a2913ff65a6877e3c064221a65bb6fc41a06f8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 11:59:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1d09f3affede4144865ff26a347a2ebf\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e4ed28ca34a4503a75c258190d11165 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a2270b78ccd4e69b52ca97e93829851 : SUCCESS in 2m 53s","accounts_in_message":[],"_revision_number":1},{"id":"0a910a2a0b0c0b955cce78ec729fc4f72883752b","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-06-23 13:57:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"edb21e94e02d806a959a1f6cda8b91ab5fe07b3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 13:57:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ace0f0f5c25c04ea80f9215ce189042c079e0eae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 14:22:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16489b6463441db962cf77c079dec6f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc9c4971c0f4b53b85d0accc4c263fb : SUCCESS in 4m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fdd5b9b2bf949b5ba291661b1af2287 : SUCCESS in 3m 26s","accounts_in_message":[],"_revision_number":1},{"id":"8c444365b2fc6ba1b49c6adf649da63a28db516a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 14:22:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1c9bc18acd0978545a4ea982dc5d884ad95062d8","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-23 14:23:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/845323fd7b714969b063ac22f2118b85\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/c83f36f5e8d946c2a9802a1c32778723 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d3fe02c9e2c08389c87d90463065cfebe937aed1","revisions":{"d3fe02c9e2c08389c87d90463065cfebe937aed1":{"kind":"REWORK","_number":1,"created":"2023-09-26 11:49:51.000000000","uploader":{"_account_id":31412,"name":"Wenping Song","email":"songwenping@inspur.com","username":"songwenping"},"ref":"refs/changes/76/896076/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/76/896076/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/76/896076/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/76/896076/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/76/896076/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/76/896076/1"}}},"commit":{"parents":[{"commit":"08886300a2cf6c2384c764caa7dd28490f071821","subject":"Maintain and update approved bp disable-enable-device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/08886300a2cf6c2384c764caa7dd28490f071821"}]}],"author":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-09-26 11:49:29.000000000","tz":480},"committer":{"name":"songwenping","email":"songwenping@inspur.com","date":"2023-09-26 11:49:29.000000000","tz":480},"subject":"move attribute spec to impletment","message":"move attribute spec to impletment\n\nChange-Id: Iea588f6b9499c35e9a60f4a2012cf6443b5e7586\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/d3fe02c9e2c08389c87d90463065cfebe937aed1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/d3fe02c9e2c08389c87d90463065cfebe937aed1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
