)]}'
{"id":"openstack%2Fcyborg-specs~978059","triplet_id":"openstack%2Fcyborg-specs~master~I170bab0a75972af426032b276a0cd46cda67e70e","project":"openstack/cyborg-specs","branch":"master","attention_set":{},"removed_from_attention_set":{"34452":{"account":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"last_update":"2026-03-02 18:48:25.000000000","reason":"Change was submitted"},"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-03-02 18:40:27.000000000","reason":"\u003cGERRIT_ACCOUNT_11604\u003e replied on the change","reason_account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}},"37598":{"account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"last_update":"2026-02-26 16:48:44.000000000","reason":"\u003cGERRIT_ACCOUNT_37598\u003e replied on the change","reason_account":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}}},"hashtags":[],"change_id":"I170bab0a75972af426032b276a0cd46cda67e70e","subject":"Add 2026.2 spec structure","status":"MERGED","created":"2026-02-26 10:40:15.000000000","updated":"2026-03-02 18:50:27.000000000","submitted":"2026-03-02 18:48:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"978059","meta_rev_id":"f203147b02f2e20e6b832c366ed05d4310b32401","_number":978059,"virtual_id_number":978059,"owner":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-02 18:48:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":2,"date":"2026-03-02 18:40:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2026-02-26 16:48:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"all":[{"value":1,"date":"2026-03-02 18:40:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}],"CC":[{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-02-26 10:40:23.000000000","updated_by":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"reviewer":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"state":"CC"},{"updated":"2026-02-26 10:50:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-02-26 16:06:10.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"state":"REVIEWER"},{"updated":"2026-02-26 16:10:47.000000000","updated_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"reviewer":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"state":"REVIEWER"}],"messages":[{"id":"a224e3dcd12103e1335a66cbfee8a14f782bfdb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 10:40:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9dff8644d07b4df5fc1ffde76ae7bb19e1757b0b","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 10:40:23.000000000","message":"Patch Set 1:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":1},{"id":"72890cc34843b5fa6279c464bb5bfadfbc81efa5","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 10:49:31.000000000","message":"Patch Set 1:\n\n(4 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/bfd6011df32f483895e197b64387e21f\n\n- teim-code-review https://zuul.teim.app/t/main/build/3fb7410c4cfa4cfaa36aea500fa0dcde : SUCCESS in 9m 00s\n\nWarning:\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: Cannot read blob 8d3b150ff91f628b5ba272afe8053d1b25bbc677\u0027\n  Comments left for invalid file specs/2026.2","accounts_in_message":[],"_revision_number":1},{"id":"48fcf0f0b61f7975b762e8eb0a60fe2b8952de61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 10:50:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/463c3a5e98b84803ad99e9db4b85b8d6\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e8d781c7347c4d1ba2f13e73f0336d52 : FAILURE in 3m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/672675f76785484faf5e0833d852fe20 : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":1},{"id":"4e09e2abe0c8df1fd9c93c06158c57ef728bc4bb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 10:52:00.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"10b4e7dca1cb6205ff889b8f06666c66233f1882","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 10:52:06.000000000","message":"Patch Set 2:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":2},{"id":"65a714557ec0937b4a8379f92704f923bba49c98","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 11:01:16.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5b8bae6306b44911b223b089d6aba229\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3fe193622cf148eca95b7e07c89ae05e : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/92937a3c1d19410e8cc81a608ac9fca2 : SUCCESS in 2m 46s","accounts_in_message":[],"_revision_number":2},{"id":"5ed7c8c3657aa6a5e47a345e88a9c5d6a74958c6","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 11:02:11.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/5d863724e9334c009ad18cff12ec6516\n\n- teim-code-review https://zuul.teim.app/t/main/build/a90ef826990f4feabf7a7b42279bd6bd : SUCCESS in 9m 57s\n\nWarning:\n  Job teim-code-review: unable to map line for file comments: \n  stderr: \u0027fatal: There is no path commit in the commit\u0027\n  Comments left for invalid file commit","accounts_in_message":[],"_revision_number":2},{"id":"e096a3c2b9ae5b63a643d4d68953e0cb1c2ef8c7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 12:01:11.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ca5326242e539a73f0a60894b13c363a6fac3adb","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 12:01:19.000000000","message":"Patch Set 3:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":3},{"id":"de0d5a2851cde7ea239b7b62079fc8dfd67bb3a9","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 12:09:11.000000000","message":"Patch Set 3:\n\n(3 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/d253a1e54fc44be99dae99ea9c650aa5\n\n- teim-code-review https://zuul.teim.app/t/main/build/cf9d09112df54c94ac0e9d4fa7e21b33 : SUCCESS in 7m 44s","accounts_in_message":[],"_revision_number":3},{"id":"88d8bdf2362289094822dabb2134a4e647b2e0c6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 12:13:15.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6544727068e44190af629ae2d41c3efd\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5c5a34e511846bfb1c6a783e6ec460d : FAILURE in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4243ded50f874d6692ca661865af3976 : SUCCESS in 2m 51s","accounts_in_message":[],"_revision_number":3},{"id":"2f289ff94c889722624ad36581291f70980eafe1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 12:53:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"3de3df784f27d2d996df65977a7507aad97f820f","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 12:53:59.000000000","message":"Patch Set 4:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":4},{"id":"537d964ed0458b3568eb83f8cb487f104cec7633","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 13:02:28.000000000","message":"Patch Set 4:\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/be1fb10dc23346cea9401e784e4d6073\n\n- teim-code-review https://zuul.teim.app/t/main/build/328356d73a2849ecacf51383f9c65f0b : SUCCESS in 8m 21s","accounts_in_message":[],"_revision_number":4},{"id":"e0977450e179fd647c6418b4a955aaca1ced00de","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 13:09:33.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d85e4b81318413d857727b4b977051f\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce799133ff08493c9f5d1866bb10b433 : SUCCESS in 4m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd163ef0ba4c4298b9a2c7eaa9f5af91 : SUCCESS in 3m 11s","accounts_in_message":[],"_revision_number":4},{"id":"711b233c91ecaaf777c0b8b9cc8b0375a8c4e7ad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 14:30:50.000000000","message":"Uploaded patch set 5: Patch Set 4 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"e2570843a3e8d58ae1edd06ed283fab5f98b63e4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 14:31:08.000000000","message":"Patch Set 5:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":5},{"id":"f5dab03194177fe8c144310aac017c618a35c607","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 14:39:51.000000000","message":"Patch Set 5:\n\n(1 comment)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/6144ee7cd0e54f429ad8604549646e1e\n\n- teim-code-review https://zuul.teim.app/t/main/build/a7f626d9cb834c44baae2e57e4fc11be : SUCCESS in 8m 32s","accounts_in_message":[],"_revision_number":5},{"id":"74e4edd60be7d875d40485330b6c037833604150","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 14:40:47.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/815e9e26516147d389045539f2844b5e\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97207fd9195d49dd9c7ef25668a0e971 : SUCCESS in 3m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/364e8040ee7e4d509c256f5ebba5de9a : SUCCESS in 3m 09s","accounts_in_message":[],"_revision_number":5},{"id":"3b93f1117e7dc1a4d8fb45232eb50e3cc98b46f5","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-26 16:06:10.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"beb461c77fe47eda4fd1afaa7ac0fa7689f1470b","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 16:10:47.000000000","message":"Patch Set 5: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":5},{"id":"a7561927362509017eacd92d691bbcfb1cfc83d1","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-02-26 16:41:02.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"f6d074273145326680ac5f182cab25d2ce0cff48","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 16:43:37.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"ac644d44acabf1f52c0f44cd9962e46e04b26f75","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 16:43:44.000000000","message":"Patch Set 6:\n\nStarting automatic-ci jobs.","accounts_in_message":[],"_revision_number":6},{"id":"d1858482689aaa54b80fc563c9e3d344fe9aa28f","author":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"date":"2026-02-26 16:43:49.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"fa2f8b79ad9037193c5cc94301fb984f6bd3e6e6","author":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"},"date":"2026-02-26 16:48:44.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"90e2dad7c94c5187815a8a68185e559911624239","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-26 16:51:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7007e0536f19453296732b8b18931f60\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6762c266033649279737163a052575dd : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17fdfcef170148d8afe955d57b110bff : SUCCESS in 4m 29s","accounts_in_message":[],"_revision_number":6},{"id":"c61a6fc187406f5cc1c71079138e4430722064c4","tag":"autogenerated:zuul:automatic-ci","author":{"_account_id":28006,"name":"teim-ci","display_name":"teim-ci","email":"ci@seanmooney.info","username":"ci-sean-mooney","status":"this is a third-party ci account run by sean-k-mooney on irc\nhosted at zuul.teim.app"},"date":"2026-02-26 16:52:24.000000000","message":"Patch Set 6:\n\n(2 comments)\n\nBuild succeeded (automatic-ci pipeline).\nhttps://zuul.teim.app/t/main/buildset/e5da481f37aa4355be0ca5e2883f0a4f\n\n- teim-code-review https://zuul.teim.app/t/main/build/c2b6a9512aa543f88daa754f18df671f : SUCCESS in 8m 31s","accounts_in_message":[],"_revision_number":6},{"id":"3b583051534892f01db427c3bcc8e8c93c6cca87","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-03-02 18:40:27.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"4d5ee09bcd4d269717ed9390a426811189abaae5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:41:13.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"9c500973a5a359b8e05a01059b5b86d4dff993cd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:48:25.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29d2b778275e4f43af9021f3fa43da55\n\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/162d508caedb4defa3bdb935ad044f1a : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc88e92100eb4395a944805b8d8c285c : SUCCESS in 1m 43s","accounts_in_message":[],"_revision_number":6},{"id":"57f54201ce8da4657a1163401c86503e218083aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:48:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"f203147b02f2e20e6b832c366ed05d4310b32401","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-02 18:50:27.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7f2ca810a044494f866e72fc48b3ed91\n\n- promote-openstack-specs https://zuul.opendev.org/t/openstack/build/aa4a169ac1cc4b2ab905dd6b519f24f0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"6416faff7234f6793e66f0d928a69c569ccceb9c","revisions":{"79ede3866fb24dd1a2ec89d8a0a06904a6141bed":{"kind":"REWORK","_number":1,"created":"2026-02-26 10:40:15.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/59/978059/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/59/978059/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/1"}}},"commit":{"parents":[{"commit":"08886300a2cf6c2384c764caa7dd28490f071821","subject":"Maintain and update approved bp disable-enable-device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/08886300a2cf6c2384c764caa7dd28490f071821"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:39:48.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:40:05.000000000","tz":60},"subject":"Add 2026.2 spec structure","message":"Add 2026.2 spec structure\n\nAdd the 2026.2 directories for specs.\n\nChange-Id: I170bab0a75972af426032b276a0cd46cda67e70e\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/79ede3866fb24dd1a2ec89d8a0a06904a6141bed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/79ede3866fb24dd1a2ec89d8a0a06904a6141bed"}]},"branch":"refs/heads/master"},"b6f17278a6219b515d53ae46d83fb37d5c86599d":{"kind":"REWORK","_number":2,"created":"2026-02-26 10:52:00.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/59/978059/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/59/978059/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/2"}}},"commit":{"parents":[{"commit":"08886300a2cf6c2384c764caa7dd28490f071821","subject":"Maintain and update approved bp disable-enable-device","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/08886300a2cf6c2384c764caa7dd28490f071821"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:39:48.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:51:55.000000000","tz":60},"subject":"Add 2026.2 spec structure","message":"Add 2026.2 spec structure\n\nAdd the 2026.2 directories for specs.\n\nChange-Id: I170bab0a75972af426032b276a0cd46cda67e70e\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b6f17278a6219b515d53ae46d83fb37d5c86599d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/b6f17278a6219b515d53ae46d83fb37d5c86599d"}]},"branch":"refs/heads/master"},"037c76be1b408c8574fad60736274b037c5e9587":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2026-02-26 12:01:11.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/59/978059/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/59/978059/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/3"}}},"commit":{"parents":[{"commit":"9433e61f01f432f049afc13479c978604790f7c8","subject":"Remove sphinxcontrib.seqdiag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9433e61f01f432f049afc13479c978604790f7c8"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:39:48.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 12:00:59.000000000","tz":60},"subject":"Add 2026.2 spec structure","message":"Add 2026.2 spec structure\n\nAdd the 2026.2 directories for specs.\n\nChange-Id: I170bab0a75972af426032b276a0cd46cda67e70e\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/037c76be1b408c8574fad60736274b037c5e9587"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/037c76be1b408c8574fad60736274b037c5e9587"}]},"branch":"refs/heads/master"},"a7bb5e483c238f3bf3da142754e07239d5f9204d":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2026-02-26 12:53:50.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/59/978059/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/59/978059/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/4"}}},"commit":{"parents":[{"commit":"9edd8e7c861b5f9e59e4f0e38a99d6374fb7d5b7","subject":"Remove sphinxcontrib.seqdiag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/9edd8e7c861b5f9e59e4f0e38a99d6374fb7d5b7"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:39:48.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 12:52:57.000000000","tz":60},"subject":"Add 2026.2 spec structure","message":"Add 2026.2 spec structure\n\nAdd the 2026.2 directories for specs.\n\nChange-Id: I170bab0a75972af426032b276a0cd46cda67e70e\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a7bb5e483c238f3bf3da142754e07239d5f9204d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/a7bb5e483c238f3bf3da142754e07239d5f9204d"}]},"branch":"refs/heads/master"},"159d3969188c64810cec13dfd2167d8214263944":{"kind":"TRIVIAL_REBASE","_number":5,"created":"2026-02-26 14:30:50.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/59/978059/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/59/978059/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/5"}}},"commit":{"parents":[{"commit":"c4a7d7833823105744f3134a16c850f9653ba68c","subject":"Remove sphinxcontrib.seqdiag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/c4a7d7833823105744f3134a16c850f9653ba68c"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:39:48.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 14:29:25.000000000","tz":60},"subject":"Add 2026.2 spec structure","message":"Add 2026.2 spec structure\n\nAdd the 2026.2 directories for specs.\n\nChange-Id: I170bab0a75972af426032b276a0cd46cda67e70e\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/159d3969188c64810cec13dfd2167d8214263944"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/159d3969188c64810cec13dfd2167d8214263944"}]},"branch":"refs/heads/master"},"6416faff7234f6793e66f0d928a69c569ccceb9c":{"kind":"REWORK","_number":6,"created":"2026-02-26 16:43:37.000000000","uploader":{"_account_id":34452,"name":"Joan Gilabert","display_name":"jgilaber","email":"jgilaber@redhat.com","username":"jgilaber"},"ref":"refs/changes/59/978059/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/cyborg-specs","ref":"refs/changes/59/978059/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/cyborg-specs refs/changes/59/978059/6"}}},"commit":{"parents":[{"commit":"c4a7d7833823105744f3134a16c850f9653ba68c","subject":"Remove sphinxcontrib.seqdiag","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/c4a7d7833823105744f3134a16c850f9653ba68c"}]}],"author":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 10:39:48.000000000","tz":60},"committer":{"name":"jgilaber","email":"jgilaber@redhat.com","date":"2026-02-26 16:43:23.000000000","tz":60},"subject":"Add 2026.2 spec structure","message":"Add 2026.2 spec structure\n\nAdd the 2026.2 directories for specs.\n\nChange-Id: I170bab0a75972af426032b276a0cd46cda67e70e\nSigned-off-by: jgilaber \u003cjgilaber@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6416faff7234f6793e66f0d928a69c569ccceb9c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/cyborg-specs/commit/6416faff7234f6793e66f0d928a69c569ccceb9c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":37598,"name":"Ivan Anfimov","display_name":"Ivan Anfimov","email":"lazekteam@gmail.com","username":"anfimovir"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
